)]}'
{"id":"openstack%2Fproject-config~797544","triplet_id":"openstack%2Fproject-config~master~Iac977829df94d5defaaf87441c6ecc60c676b16c","project":"openstack/project-config","branch":"master","topic":"gating.dev","hashtags":[],"change_id":"Iac977829df94d5defaaf87441c6ecc60c676b16c","subject":"Adding gating.dev zone repo","status":"MERGED","created":"2021-06-22 22:06:43.000000000","updated":"2021-06-22 23:59:14.000000000","submitted":"2021-06-22 23:31:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797544-gating.dev","meta_rev_id":"fce04dd93ee8ed2220238ef778f21956de41f8c6","_number":797544,"virtual_id_number":797544,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-22 23:31:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-22 23:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2021-06-22 22:20:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-22 23:20:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-22 22:14:06.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-06-22 22:14:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-22 23:20:02.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"0382474858107b2f0659bf507e4087c9a1646315","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2021-06-22 22:06:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7d32e6c677930c5b0991d01d53744135ffc8b1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2021-06-22 22:08:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7a24d290afd83b82347f8f363b190375d9cacbdd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2021-06-22 22:11:33.000000000","message":"Topic set to gating.dev","accounts_in_message":[],"_revision_number":2},{"id":"1543faa98f3214978a30b69282f3e4a4a6079f20","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-06-22 22:14:06.000000000","message":"Patch Set 2: Code-Review+2\n\nAn alternative would be to reuse zone-zuul-ci.org since it already contains to domains (zuulci.org being the other one).","accounts_in_message":[],"_revision_number":2},{"id":"d3f4ac7ac20e36a09c9ce8c60d96b97ef27f151a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 22:14:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6318144d1c9f4d258f2e0d1bcca3ab91 : SUCCESS in 5m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/66e9ba18ebfd4aaaabe5b42a63b1b23f : FAILURE in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"c774929f88ac71ae1964c0ef6587cd6fb1ea0b9f","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2021-06-22 22:18:09.000000000","message":"Patch Set 2:\n\nTrue.  I\u0027m sort of ambivalent about that with a slight preference for this since it\u0027s a distinctly different domain name than the name of the zuul-ci.org zone repo.","accounts_in_message":[],"_revision_number":2},{"id":"1ffac929d7a1388cd7eb72748a76a1e57923bf6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2021-06-22 22:18:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"32094de0983a5ba95dc92121b31f17b50dab9ba2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-06-22 22:20:05.000000000","message":"Patch Set 3: Code-Review+2\n\nYep, I agree, it\u0027ll be different site content not merely an alias/redirect.","accounts_in_message":[],"_revision_number":3},{"id":"7cc64011e7b58dc194535739830d65ea699f6433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 22:26:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5b8ebaedfe444ac59f72978894067cff : SUCCESS in 5m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/98b8cb95163044cd885a08a5fff3eda6 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":3},{"id":"64af6cb042591cf54e7d140ecbfe87597e13af8d","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-22 23:20:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"727495314b0157a9a72d5c30a30761ab0cc6a689","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 23:20:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1e7418fc7fa4b84292526bc5bbd54e1b6aea9a48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 23:31:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b2f5da2b23d64999b41cb544cdc6762d : SUCCESS in 5m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/633659a9d7834e3fa37334f2d8309492 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":3},{"id":"3e23b0ce69d97ff5e7f01120a4829ae767452d39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 23:31:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fce04dd93ee8ed2220238ef778f21956de41f8c6","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 23:59:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/e33eb186ba414ffea8ce0dc6685ce9c4 : SUCCESS in 2m 35s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/705706fd5fb14d61832eafc8fbe7592a : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4a8cb120844168eaf6c2126a495d9b7b328d98eb","revisions":{"7e9f5de46531c19ff4323f3f7a3ee3bc03159a37":{"kind":"REWORK","_number":1,"created":"2021-06-22 22:06:43.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/44/797544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/44/797544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/797544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/797544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/797544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/44/797544/1"}}},"commit":{"parents":[{"commit":"1ddf30fab14607a41c5decc66720d82d3cef1637","subject":"Add zuul/gating.dev repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1ddf30fab14607a41c5decc66720d82d3cef1637"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2021-06-22 22:06:19.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2021-06-22 22:06:19.000000000","tz":-420},"subject":"Adding gating.dev zone repo","message":"Adding gating.dev zone repo\n\nChange-Id: Iac977829df94d5defaaf87441c6ecc60c676b16c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e9f5de46531c19ff4323f3f7a3ee3bc03159a37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e9f5de46531c19ff4323f3f7a3ee3bc03159a37"}]},"branch":"refs/heads/master"},"f2c2f6fa26d57f4d1bb843717eff10459eab68eb":{"kind":"REWORK","_number":2,"created":"2021-06-22 22:08:18.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/44/797544/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/44/797544/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/797544/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/797544/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/797544/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/44/797544/2"}}},"commit":{"parents":[{"commit":"1ddf30fab14607a41c5decc66720d82d3cef1637","subject":"Add zuul/gating.dev repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1ddf30fab14607a41c5decc66720d82d3cef1637"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2021-06-22 22:06:19.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2021-06-22 22:08:12.000000000","tz":-420},"subject":"Adding gating.dev zone repo","message":"Adding gating.dev zone repo\n\nChange-Id: Iac977829df94d5defaaf87441c6ecc60c676b16c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2c2f6fa26d57f4d1bb843717eff10459eab68eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2c2f6fa26d57f4d1bb843717eff10459eab68eb"}]},"branch":"refs/heads/master"},"4a8cb120844168eaf6c2126a495d9b7b328d98eb":{"kind":"REWORK","_number":3,"created":"2021-06-22 22:18:18.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/44/797544/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/44/797544/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/797544/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/797544/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/797544/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/44/797544/3"}}},"commit":{"parents":[{"commit":"1ddf30fab14607a41c5decc66720d82d3cef1637","subject":"Add zuul/gating.dev repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1ddf30fab14607a41c5decc66720d82d3cef1637"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2021-06-22 22:06:19.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2021-06-22 22:16:52.000000000","tz":-420},"subject":"Adding gating.dev zone repo","message":"Adding gating.dev zone repo\n\nChange-Id: Iac977829df94d5defaaf87441c6ecc60c676b16c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a8cb120844168eaf6c2126a495d9b7b328d98eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a8cb120844168eaf6c2126a495d9b7b328d98eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
