)]}'
{"id":"openstack%2Fproject-config~797747","triplet_id":"openstack%2Fproject-config~master~I541bd96bf573db7d138ccedb4c3b83f359b21fe9","project":"openstack/project-config","branch":"master","topic":"retire-scientific-wg","hashtags":[],"change_id":"I541bd96bf573db7d138ccedb4c3b83f359b21fe9","subject":"Remove scientific-wg from infra","status":"MERGED","created":"2021-06-23 20:38:52.000000000","updated":"2021-06-29 02:30:23.000000000","submitted":"2021-06-29 02:11:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797747-retire-scientific-wg","meta_rev_id":"9244b5587f900631da7ffcf718f5fb1705516a6e","_number":797747,"virtual_id_number":797747,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-29 02:11:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-29 00:27:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-29 00:27:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 20:46:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-29 00:27:36.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"b3edb031b1cd6a45d85c926f106ff2cdf6025cdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-23 20:38:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e10b046a4391cd2563ca7984f4d11691aaa1d52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 20:46:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ea186b21848f40d2a8990028af881f75 : SUCCESS in 5m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ec37e36dbc6045a2b623f0cd39393560 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"cc1bb19d02d3e91c99358e1cd523798a84551663","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-28 22:12:12.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"acfd7d238f7e1bb9d0e4bde3808cc5d7e220b345","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 22:25:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3be36808516f41e4871d7ffac7a9d3c4 : SUCCESS in 7m 37s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1b36633758044d0dab6d5366ab501706 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":2},{"id":"4f8f39dee2339be435ae4fe8c8d131ecce6b1501","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-06-29 00:27:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"280e2ff649d1b4bc6ce81c0df063bbdf221529cc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-29 01:45:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"81f29931cae768a3b3111201fa627112a97ce622","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 01:57:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/460afcbf0be4437d83aa549bcea3bc05 : SUCCESS in 7m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/936a8e70e0af432299f0c73e70542ee9 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":2},{"id":"918f65fdcb828f348eee0ab9ee868b65c7b203de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 01:58:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6361cfb8c44ab0e8799267b32beaaafb32081c3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 02:11:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d39e3e3547bd4fc4be6d32bf37f0b559 : SUCCESS in 7m 38s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7ae868c06b4a40f49a759723bc5d2570 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":2},{"id":"3c043409031ee35fbf1d1ed147fff31d45197c42","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 02:11:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9244b5587f900631da7ffcf718f5fb1705516a6e","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 02:30:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/2235cd62b514484589097c31b783d0d9 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f21a16c440ed4c2ca238da83e614439ce13e534","revisions":{"25bde96c974450a39de4f9e33fd48977f2aafc8a":{"kind":"REWORK","_number":1,"created":"2021-06-23 20:38:52.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/797747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/797747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/797747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/797747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/797747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/797747/1"}}},"commit":{"parents":[{"commit":"668345bbe7a4210ecd67198a3afa2a19c9424118","subject":"End project gating for retiring scientific-wg repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/668345bbe7a4210ecd67198a3afa2a19c9424118"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-06-23 20:36:25.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-06-23 20:37:56.000000000","tz":-300},"subject":"Remove scientific-wg from infra","message":"Remove scientific-wg from infra\n\nscientific-wg repo is retired, this commit remove it\nfrom infra.\n\n- https://review.opendev.org/q/topic:%22retire-scientific-wg%22+(status:open%20OR%20status:merged)\n\nDepends-On: https://review.opendev.org/c/openstack/scientific-wg/+/797746\nChange-Id: I541bd96bf573db7d138ccedb4c3b83f359b21fe9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/25bde96c974450a39de4f9e33fd48977f2aafc8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/25bde96c974450a39de4f9e33fd48977f2aafc8a"}]},"branch":"refs/heads/master"},"6f21a16c440ed4c2ca238da83e614439ce13e534":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-28 22:12:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/797747/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/797747/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/797747/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/797747/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/797747/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/797747/2"}}},"commit":{"parents":[{"commit":"f18900f48af75621d7370a5ccd65a20b4081513b","subject":"Merge \"Update Zuul job semaphore usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f18900f48af75621d7370a5ccd65a20b4081513b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-06-23 20:36:25.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-06-28 22:12:12.000000000","tz":0},"subject":"Remove scientific-wg from infra","message":"Remove scientific-wg from infra\n\nscientific-wg repo is retired, this commit remove it\nfrom infra.\n\n- https://review.opendev.org/q/topic:%22retire-scientific-wg%22+(status:open%20OR%20status:merged)\n\nDepends-On: https://review.opendev.org/c/openstack/scientific-wg/+/797746\nChange-Id: I541bd96bf573db7d138ccedb4c3b83f359b21fe9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f21a16c440ed4c2ca238da83e614439ce13e534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f21a16c440ed4c2ca238da83e614439ce13e534"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
