)]}'
{"id":"openstack%2Fproject-config~797781","triplet_id":"openstack%2Fproject-config~master~I6125a9c8985847fd957dcb83bf5ea2865bdd2672","project":"openstack/project-config","branch":"master","topic":"retire-openstack-specs","hashtags":[],"change_id":"I6125a9c8985847fd957dcb83bf5ea2865bdd2672","subject":"End proejct gating for puppet-openstack-specs","status":"ABANDONED","created":"2021-06-23 21:30:04.000000000","updated":"2021-06-28 22:29:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7454c1a2632fd81210bf62381b9941e732232c0f","_number":797781,"virtual_id_number":797781,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-23 21:36:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 21:36:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7fa9994f060a5c254a3c78ef31de1feec70c5b50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-23 21:30:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2af3ae9770f585dcef3289c1513373f3f3c65319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 21:36:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/532a390d8f0b4c61aaa13d2ac1bcc231 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7454c1a2632fd81210bf62381b9941e732232c0f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-28 22:29:29.000000000","message":"Abandoned\n\ndone in https://review.opendev.org/c/openstack/project-config/+/798392","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f80367b554ea6d24fb1abc2d425a3ac66439a70","revisions":{"4f80367b554ea6d24fb1abc2d425a3ac66439a70":{"kind":"REWORK","_number":1,"created":"2021-06-23 21:30:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/81/797781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/81/797781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/797781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/797781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/797781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/81/797781/1"}}},"commit":{"parents":[{"commit":"0a3af179f5d713503402399504e22cc8aa405091","subject":"Merge \"Add publish-to-pypi job template for tripleo-operator-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a3af179f5d713503402399504e22cc8aa405091"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-06-23 21:28:58.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-06-23 21:28:58.000000000","tz":-300},"subject":"End proejct gating for puppet-openstack-specs","message":"End proejct gating for puppet-openstack-specs\n\nopenstack-specs repo is retired now, so we can retire\npuppet-openstack-specs also:\n\nhttps: //review.opendev.org/q/topic:%22retire-openstack-specs%22+(status:open%20OR%20status:merged)\nChange-Id: I6125a9c8985847fd957dcb83bf5ea2865bdd2672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f80367b554ea6d24fb1abc2d425a3ac66439a70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f80367b554ea6d24fb1abc2d425a3ac66439a70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
