)]}'
{"id":"openstack%2Fproject-config~797784","triplet_id":"openstack%2Fproject-config~master~Id4b1a5d6d5e32e92f93b012e0fec5138d405e4b7","project":"openstack/project-config","branch":"master","topic":"retire-openstack-specs","hashtags":[],"change_id":"Id4b1a5d6d5e32e92f93b012e0fec5138d405e4b7","subject":"Remove puppet-openstack-specs from infra","status":"MERGED","created":"2021-06-23 21:38:32.000000000","updated":"2021-06-28 22:07:13.000000000","submitted":"2021-06-28 21:17:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"797784-retire-openstack-specs","meta_rev_id":"0904c6e66d3e51ed53da9fbd2e08675ffa90a17d","_number":797784,"virtual_id_number":797784,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-28 21:17:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2021-06-26 07:47:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-28 21:03:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-28 21:03:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 21:53:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-25 12:44:11.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-06-28 21:03:40.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"4c5ec94bdc37b853795e37d6d7b479152ed9a233","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-23 21:38:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fa581a170aaa2d3a5f5939932222ff4b4c9ef96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 21:53:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8abae4b8a0d348f5b8d9a1f239eda7d4 : SUCCESS in 5m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/97dcd17540b04c6d96cccb800fcb4b01 : SUCCESS in 4m 45s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/787285ccfba74d92b6b454e0f8d5844c : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e789640e32a4e289de6f7f1eb615b58c83507e58","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-06-25 12:44:11.000000000","message":"Patch Set 1: Code-Review-1\n\nremove both openstack-specs and puppet-openstack-specs from this list as well:\n\nhttps://opendev.org/openstack/project-config/src/branch/master/specs/specs.yaml#L45","accounts_in_message":[],"_revision_number":1},{"id":"18d8c650f1758b8acdc62b0a7b5a5c5a7099fd99","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-25 15:52:28.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e remove both openstack-specs and puppet-openstack-specs from this list as well:\n\u003e \n\u003e https://opendev.org/openstack/project-config/src/branch/master/specs/specs.yaml#L45\n\ndone","accounts_in_message":[],"_revision_number":1},{"id":"66ffda3e4af574c446a141e6680c80e409e210db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-25 15:52:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cf3d96f860beb05b7f729346515b9e217fdc5471","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 16:03:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/40dada1dbafd47aeacb36c9dc6b8487f : SUCCESS in 5m 39s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/60a51e1284c941d3b9b344173bb2d6cf : SUCCESS in 7m 10s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/74886d84c5374db28baa938d5aa7e3ce : SUCCESS in 6m 21s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/b982de1e650749e7b0bbf820200e335e : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":2},{"id":"49e8ca3adb91bf101d184c7c44f4fcabe894bfdd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-06-26 07:47:47.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6cb5c8e90f9d5a304a1544e5c75560f183d27fa8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-06-28 21:03:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c79768c8b44689a6673c03f912720c015a4046a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 21:04:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"11573a611931a2ef033e68e45f1d9cde587a8bd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 21:17:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/88d3847582d54cfeb35b6198af744d35 : SUCCESS in 5m 50s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c4a9e65ce5034755bbf8ebba9b394c2d : SUCCESS in 7m 23s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/59fe04fe9cc646a4845865e6362e2a3c : SUCCESS in 6m 19s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/800f59acf6164956a86e8d0d62f020a1 : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":2},{"id":"4d0112969e99fbfc0ba70ad09e26de3e06fe4206","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 21:17:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0904c6e66d3e51ed53da9fbd2e08675ffa90a17d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 22:07:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/d64a1ad6695c407c994eeca9efe26cca : SUCCESS in 1m 13s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/db560b9cc4cf425d9179e4d366e4f245 : SUCCESS in 3m 05s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/1fe2383f3c6144b5b934ed4eb6c04f5a : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"57464b6ae6c3b59f79759ee95089048fe4fd33cb","revisions":{"b1c6ba8b24a4c9614c0a039f966d8ed42830fbad":{"kind":"REWORK","_number":1,"created":"2021-06-23 21:38:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/84/797784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/84/797784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/797784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/797784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/797784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/84/797784/1"}}},"commit":{"parents":[{"commit":"2518a66f1cbd8df58bbea37fb645615600a7352b","subject":"Remove uc-recognition from infra","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2518a66f1cbd8df58bbea37fb645615600a7352b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-06-23 21:37:54.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-06-23 21:37:54.000000000","tz":-300},"subject":"Remove puppet-openstack-specs from infra","message":"Remove puppet-openstack-specs from infra\n\npuppet-openstack-specs repo is retired, this commit remove it\nfrom infra.\n\n- https://review.opendev.org/q/topic:%22retire-openstack-specs%22+(status:open%20OR%20status:merged)\n\nChange-Id: Id4b1a5d6d5e32e92f93b012e0fec5138d405e4b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b1c6ba8b24a4c9614c0a039f966d8ed42830fbad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b1c6ba8b24a4c9614c0a039f966d8ed42830fbad"}]},"branch":"refs/heads/master"},"57464b6ae6c3b59f79759ee95089048fe4fd33cb":{"kind":"REWORK","_number":2,"created":"2021-06-25 15:52:34.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/84/797784/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/84/797784/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/797784/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/797784/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/797784/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/84/797784/2"}}},"commit":{"parents":[{"commit":"ad58188df909c30d55457739c61ea9921d80ca33","subject":"Remove uc-recognition from infra","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad58188df909c30d55457739c61ea9921d80ca33"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-06-25 15:51:33.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-06-25 15:51:33.000000000","tz":-300},"subject":"Remove puppet-openstack-specs from infra","message":"Remove puppet-openstack-specs from infra\n\npuppet-openstack-specs repo is retired, this commit remove it\nfrom infra.\n\n- https://review.opendev.org/q/topic:%22retire-openstack-specs%22+(status:open%20OR%20status:merged)\n\nChange-Id: Id4b1a5d6d5e32e92f93b012e0fec5138d405e4b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57464b6ae6c3b59f79759ee95089048fe4fd33cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57464b6ae6c3b59f79759ee95089048fe4fd33cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
