)]}'
{"id":"openstack%2Fproject-config~797790","triplet_id":"openstack%2Fproject-config~master~If013a67fb01f23cb6cef9610f218fa6b55067851","project":"openstack/project-config","branch":"master","topic":"pause-failing-builds","hashtags":[],"change_id":"If013a67fb01f23cb6cef9610f218fa6b55067851","subject":"nodepool: pause gentoo and tumbleweed builds","status":"MERGED","created":"2021-06-23 23:26:22.000000000","updated":"2021-06-24 00:01:53.000000000","submitted":"2021-06-23 23:56:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797790-pause-failing-builds","meta_rev_id":"54db460c1e995a81256faaf3ecf6ed7de28336d7","_number":797790,"virtual_id_number":797790,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-23 23:55:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-23 23:46:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-23 23:46:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 23:33:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b788df20aa9709001f91c3ef51eabfd0baba2af6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-23 23:26:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7875a7dbaec45c0217cd0292234b6c2c35d3ef52","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-23 23:28:59.000000000","message":"Patch Set 1:\n\ngentoo failure : http://paste.openstack.org/show/806899/\n\ntumbleweed failure: http://paste.openstack.org/show/806900/","accounts_in_message":[],"_revision_number":1},{"id":"ad9500deaca492fde2418d6d687eda249e4e4f5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 23:33:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/18ea2d6bda694dd4bb7f959b82e8509d : SUCCESS in 6m 07s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/021d42245be348bea58ad88d4bfeae1d : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f46da35f907dd16c8ad44f2ceb8bd07c1a681bb6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-23 23:46:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nunfortunately the dib gate is currently broken with some sort of centos-8-stream grub2 hanging-during-install issue, so we can\u0027t quickly roll out any fix that would be forthcoming; and we need the maintainers of these platforms to look into this.  thus i\u0027m taking the liberty of self-approving this to stop these invalid builds constantly looping","accounts_in_message":[],"_revision_number":1},{"id":"eeacede497542f948055d665974a91d19520d5c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 23:46:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f960abd22f63fa830623d4d423ee24be5efdebe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 23:55:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f15e37b76c3f48528037c7c5cc1e6db1 : SUCCESS in 5m 42s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/a7ffe99bc3bc432eb03436d84d69550b : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"460ce6597453a5b372271e837539c9e4a118e166","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 23:56:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"54db460c1e995a81256faaf3ecf6ed7de28336d7","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 00:01:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/c4b58ce3d43842e0abd439783f94daba : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"356b63f7bcd3d3a19c032a9c81edc685a3814673","revisions":{"356b63f7bcd3d3a19c032a9c81edc685a3814673":{"kind":"REWORK","_number":1,"created":"2021-06-23 23:26:22.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/90/797790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/797790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/797790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/797790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/797790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/797790/1"}}},"commit":{"parents":[{"commit":"0a3af179f5d713503402399504e22cc8aa405091","subject":"Merge \"Add publish-to-pypi job template for tripleo-operator-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a3af179f5d713503402399504e22cc8aa405091"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-06-23 23:25:06.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-06-23 23:25:06.000000000","tz":600},"subject":"nodepool: pause gentoo and tumbleweed builds","message":"nodepool: pause gentoo and tumbleweed builds\n\nBoth of these are failing in ways that look like we need to fix them\nin dib.  Stop attempting to build them for now.\n\nChange-Id: If013a67fb01f23cb6cef9610f218fa6b55067851\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/356b63f7bcd3d3a19c032a9c81edc685a3814673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/356b63f7bcd3d3a19c032a9c81edc685a3814673"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
