)]}'
{"id":"openstack%2Fproject-config~797806","triplet_id":"openstack%2Fproject-config~master~Ieabc9d9b51e0a908480dfb56cfc5baf0d0760365","project":"openstack/project-config","branch":"master","topic":"pause-failing-builds","hashtags":[],"change_id":"Ieabc9d9b51e0a908480dfb56cfc5baf0d0760365","subject":"nodepool: pause centos-8-stream builds","status":"MERGED","created":"2021-06-24 06:32:43.000000000","updated":"2021-06-24 08:32:17.000000000","submitted":"2021-06-24 07:22:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797806-pause-failing-builds","meta_rev_id":"b20ac0e4834fcfe355e58895c5c233085a386233","_number":797806,"virtual_id_number":797806,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-24 07:22:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-24 07:04:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-24 07:04:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-24 06:44:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d54279f50b7b1f920cb13ccfebc445e8e308d14c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-24 06:32:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"709d0e94759fbdd3fbfe7392db6dee7a22cf0fd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 06:44:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0d1d797fa0704c92bea6bce40f625d3d : SUCCESS in 7m 58s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/5e6e3d08de294de1a045a3eb34597b9a : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"15e1b1e7bea596710a55250ae69e709ef96958db","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-24 07:04:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\ntaking the liberty of merging this, becuase the job times out with RPM in the middle of operations and gets killed, leaving mounts and all sorts of bad things.","accounts_in_message":[],"_revision_number":1},{"id":"d8993f6e4c4ae03a0c27d07672f1315ea5520458","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 07:05:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d21d7caaa3504bbc6fbcc4c5ad7518ea068469b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 07:22:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/aa3b2e7f5a5e43c5bc878be85d3b545e : SUCCESS in 11m 36s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/541fec766220411b9030a44fbd2f7898 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7828d0a631559660a02b6bc27650d062da05ffd8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 07:22:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b20ac0e4834fcfe355e58895c5c233085a386233","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 08:32:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/e22bef2548c3445c900dc16e5393a543 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"813542fabd0f9563283833b9d455ea070b24b79c","revisions":{"813542fabd0f9563283833b9d455ea070b24b79c":{"kind":"REWORK","_number":1,"created":"2021-06-24 06:32:43.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/06/797806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/797806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/797806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/797806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/797806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/797806/1"}}},"commit":{"parents":[{"commit":"356b63f7bcd3d3a19c032a9c81edc685a3814673","subject":"nodepool: pause gentoo and tumbleweed builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/356b63f7bcd3d3a19c032a9c81edc685a3814673"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-06-24 06:30:50.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-06-24 06:30:50.000000000","tz":600},"subject":"nodepool: pause centos-8-stream builds","message":"nodepool: pause centos-8-stream builds\n\nA new release of rng-tools is causing the build to constantly time-out\n[1].  I\u0027m not confident with the dib change until it\u0027s had more review\n[2], and it will take us some time to release or a fixed package to\ncome through anyway.  In the mean-time, pause the build.\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1975588\n[2] https://review.opendev.org/c/openstack/diskimage-builder/+/797794\n\nChange-Id: Ieabc9d9b51e0a908480dfb56cfc5baf0d0760365\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/813542fabd0f9563283833b9d455ea070b24b79c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/813542fabd0f9563283833b9d455ea070b24b79c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
