)]}'
{"id":"openstack%2Fproject-config~799123","triplet_id":"openstack%2Fproject-config~master~I597c1a577c53e2db61413d9ec531378667691d2a","project":"openstack/project-config","branch":"master","topic":"gerrit-3.2","hashtags":[],"change_id":"I597c1a577c53e2db61413d9ec531378667691d2a","subject":"Drop use of track-upstream","status":"MERGED","created":"2021-07-01 22:47:16.000000000","updated":"2021-07-06 18:46:21.000000000","submitted":"2021-07-06 18:41:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"799123-gerrit-3.2","meta_rev_id":"c6d1bada1a660af38df7dc6ceb3764e315173745","_number":799123,"virtual_id_number":799123,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-06 18:41:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-07-06 18:33:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2021-07-06 18:18:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2021-07-06 18:33:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-01 22:58:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-04 12:59:17.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"CC"},{"updated":"2021-07-06 08:33:11.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-07-06 15:39:54.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"80a1027afb6212e0a4067f22f54b3354bc157132","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-07-01 22:47:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d5ae060ada41fd1fbf75910675039d944c4337c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 22:58:30.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a492c618eb324e49bce3e819e3bd12fe : FAILURE in 5m 08s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/68b2d231d0b540538574bd1226340fa2 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7161bd70fb40162f9f44648e75e93af3d687b200","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-07-01 23:14:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"224087944582d1f9205c2dae827456028eed214a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 23:23:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/b28cfbdfabd943a483b9d4e1bf6e20c3 : SUCCESS in 4m 34s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/17f7ff0ef84e482688ee81973a5a84df : SUCCESS in 6m 06s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/9bf327aecd9d4cdb943f7fc3ff9d1459 : SUCCESS in 3m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e5d59f6a6e204faf807a6387d825950a : SUCCESS in 6m 01s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/1fa36d267bf0463bb5e4c3dd898f7259 : SUCCESS in 4m 34s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/370ce734abab4f27a095cb79f1c54a4b : SUCCESS in 4m 06s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/9dc6d108ff9a48b69bcf2ed37f41795a : SUCCESS in 6m 41s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/34b27251a3c74de99683a23101f96922 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d3ae908371fa671b76f98e3e6ba50f55c27ae42c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-04 12:59:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af632405e0bc6fa481df893bb2293c202734cb75","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-07-04 14:32:40.000000000","message":"Patch Set 2: Workflow-1\n\nAndreas Jaeger wrote:\n\n\u003e (1 comment)\n\nOoh, yep great catch. Thanks!","accounts_in_message":[],"_revision_number":2},{"id":"8550fce39b1cdafad5aa63bca5be19f5f82319bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-07-04 14:34:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"020b9c0405c011d6d9121faaef7a7da476b78929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-04 14:43:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/468342764189405a899916cb72e2dd1f : SUCCESS in 4m 35s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3e48c713895f4b2a96a6a22381625b99 : SUCCESS in 6m 56s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/0daea0f6de3f4a99a16b3db5c47b6502 : SUCCESS in 4m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ee768c2095424f468095542df7df797b : SUCCESS in 5m 40s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/0ba69e4739d640dbb364ef59c5fa2e39 : SUCCESS in 4m 23s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/d00c3961b6224659a245989305f85e52 : SUCCESS in 6m 00s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/32fd31262af74c2084b745d89d0bd6df : SUCCESS in 6m 52s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/78af3f21a5bd425aa6af56e70dcad39b : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":3},{"id":"247c7b5e5f5fa626770f0b8fbd1c4eb644feb5ae","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-06 08:33:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"789bc42c3b827203bf58db63d1d6c24c6da4b91f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-07-06 15:39:54.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3bce5d7e546b62d2c5e0b69722370d5fa8805bf7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-07-06 16:44:14.000000000","message":"Patch Set 3: Workflow-1\n\nYep, great suggestion, thanks!","accounts_in_message":[],"_revision_number":3},{"id":"415fa27094188f4c93c459fc0c45131ff71896ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-07-06 16:45:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6353974ecf5c8f9e5f019cbf10f5d87e3e10ac56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 16:56:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/57a411d24edd41f3b7985bb510f5c666 : SUCCESS in 4m 07s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8f00f8ebfb2147a59e61e01252846c18 : SUCCESS in 7m 43s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/45ea008c2b984b3ea8e7d9c28b9f56c2 : SUCCESS in 5m 26s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/80f118bc674d4e7a988ca91c5d236643 : SUCCESS in 5m 54s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/878a2d12fce347ebb2085184640e5dba : SUCCESS in 5m 44s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/a1786315bee64f48b53166be0296fad2 : SUCCESS in 5m 37s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/0ad616e20c544687962898f9a32e2c35 : SUCCESS in 7m 54s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/00c0e39c65fe4583bac2eee08a915cb6 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":4},{"id":"77e818ecaa0e3e1b5a3be71ab574e846f97bb4af","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-06 18:18:44.000000000","message":"Patch Set 4: Code-Review+2\n\nCan we drop opendev/gerrit completely now or is it still needed?","accounts_in_message":[],"_revision_number":4},{"id":"c9cdb462685f6f898d146301c6606434515f977b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-07-06 18:33:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0b4004baeaab2b133792c65f50a5d3678f64648a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 18:33:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2ffa7587eff6bca8635caa6ba53188b292d64e0f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-07-06 18:40:57.000000000","message":"Patch Set 4:\n\nAndreas Jaeger wrote:\n\n\u003e Can we drop opendev/gerrit completely now or is it still needed?\n\nI figured a followup step is to retire that repository, yes. We\u0027ve been installing from images and patching the upstream source on the fly in our image builds for a while, though more recently we were able to drop the patches. Regardless, I think if we do need to fork Gerrit again we\u0027d more likely go back to patching the source at image build time rather than building from a tracked fork of their Git repository in our Gerrit deployment.","accounts_in_message":[],"_revision_number":4},{"id":"040bb775180b631567ed13e9a4dfe0b4f18f256e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 18:41:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/dde329b10d9947bba9c7e9d57a517575 : SUCCESS in 5m 22s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9fba246f1fd34a9c818669ca11f4a7cc : SUCCESS in 6m 42s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/13a1faffc5874a3da32d9fbf1a936829 : SUCCESS in 4m 06s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/adb41fa99df8447385748890a143eefd : SUCCESS in 4m 34s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/03bb2afbfd444d8a8f32923348600fa3 : SUCCESS in 4m 38s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/d2fb67d979de43bdb309d9053e4f29bc : SUCCESS in 4m 11s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/3ead9de6e03b46a9bf3a7aeeb4e6bf99 : SUCCESS in 6m 43s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/4ea15516fa17449ab0e88d0f93e9c53a : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":4},{"id":"e8cb806d35c7a32413020ec6c3e67ee4b94b323a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 18:41:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c6d1bada1a660af38df7dc6ceb3764e315173745","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 18:46:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\n\n- promote-infra-index https://zuul.opendev.org/t/openstack/build/8126b176f8f3442a87dd28e96f74804b : SUCCESS in 59s\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/dcb125e69a08420d9e7618e639cf030a : SUCCESS in 1m 05s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/b1d3b92e318841a5b217b86a1f2a5fc9 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ef03e7fa36e0de5f89abf3bdea10fdd8eea151d6","revisions":{"33f33b5fd1497219bb13c808a959d3d1cb2de308":{"kind":"REWORK","_number":1,"created":"2021-07-01 22:47:16.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/23/799123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/799123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/799123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/799123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/799123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/799123/1"}}},"commit":{"parents":[{"commit":"143922e7103866e4457a35255b569e3667f98e93","subject":"Merge \"Correct branch path in afsdocs_secret-releasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/143922e7103866e4457a35255b569e3667f98e93"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-07-01 22:45:22.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-07-01 22:45:22.000000000","tz":0},"subject":"Drop use of track-upstream","message":"Drop use of track-upstream\n\nWhen upgrading from Gerrit 2.13 to 3.2 we stopped relying on a local\nfork of its source and have been building from (more recently\ncompletely unmolested) upstream source. This fork was the only place\nwe were relying on jeepyb\u0027s track-upstream feature, so we can stop\nlooking for it in our checks and normalization as well.\n\nChange-Id: I597c1a577c53e2db61413d9ec531378667691d2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/33f33b5fd1497219bb13c808a959d3d1cb2de308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/33f33b5fd1497219bb13c808a959d3d1cb2de308"}]},"branch":"refs/heads/master"},"8226436c9cbd5eb9015ceb181c99e85a264e9418":{"kind":"REWORK","_number":2,"created":"2021-07-01 23:14:00.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/23/799123/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/799123/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/799123/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/799123/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/799123/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/799123/2"}}},"commit":{"parents":[{"commit":"143922e7103866e4457a35255b569e3667f98e93","subject":"Merge \"Correct branch path in afsdocs_secret-releasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/143922e7103866e4457a35255b569e3667f98e93"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-07-01 22:45:22.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-07-01 23:11:55.000000000","tz":0},"subject":"Drop use of track-upstream","message":"Drop use of track-upstream\n\nWhen upgrading from Gerrit 2.13 to 3.2 we stopped relying on a local\nfork of its source and have been building from (more recently\ncompletely unmolested) upstream source. This fork was the only place\nwe were relying on jeepyb\u0027s track-upstream feature, so we can stop\nlooking for it in our checks and normalization as well.\n\nRemove the check_gerrit_projects_changed scripts as well, as they\nonly existed to run things where track-upstream was enabled.\n\nChange-Id: I597c1a577c53e2db61413d9ec531378667691d2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8226436c9cbd5eb9015ceb181c99e85a264e9418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8226436c9cbd5eb9015ceb181c99e85a264e9418"}]},"branch":"refs/heads/master"},"f4184807050f20dc9d49aa5258d94aaf79647de9":{"kind":"REWORK","_number":3,"created":"2021-07-04 14:34:01.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/23/799123/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/799123/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/799123/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/799123/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/799123/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/799123/3"}}},"commit":{"parents":[{"commit":"143922e7103866e4457a35255b569e3667f98e93","subject":"Merge \"Correct branch path in afsdocs_secret-releasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/143922e7103866e4457a35255b569e3667f98e93"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-07-01 22:45:22.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-07-04 14:33:52.000000000","tz":0},"subject":"Drop use of track-upstream","message":"Drop use of track-upstream\n\nWhen upgrading from Gerrit 2.13 to 3.2 we stopped relying on a local\nfork of its source and have been building from (more recently\ncompletely unmolested) upstream source. This fork was the only place\nwe were relying on jeepyb\u0027s track-upstream feature, so we can stop\nlooking for it in our checks and normalization as well.\n\nRemove the check_gerrit_projects_changed scripts as well, as they\nonly existed to run things where track-upstream was enabled.\n\nChange-Id: I597c1a577c53e2db61413d9ec531378667691d2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f4184807050f20dc9d49aa5258d94aaf79647de9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f4184807050f20dc9d49aa5258d94aaf79647de9"}]},"branch":"refs/heads/master"},"ef03e7fa36e0de5f89abf3bdea10fdd8eea151d6":{"kind":"REWORK","_number":4,"created":"2021-07-06 16:45:30.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/23/799123/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/799123/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/799123/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/799123/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/799123/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/799123/4"}}},"commit":{"parents":[{"commit":"143922e7103866e4457a35255b569e3667f98e93","subject":"Merge \"Correct branch path in afsdocs_secret-releasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/143922e7103866e4457a35255b569e3667f98e93"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-07-01 22:45:22.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-07-06 16:45:22.000000000","tz":0},"subject":"Drop use of track-upstream","message":"Drop use of track-upstream\n\nWhen upgrading from Gerrit 2.13 to 3.2 we stopped relying on a local\nfork of its source and have been building from (more recently\ncompletely unmolested) upstream source. This fork was the only place\nwe were relying on jeepyb\u0027s track-upstream feature, so we can stop\nlooking for it in our checks and normalization as well.\n\nRemove the check_gerrit_projects_changed scripts as well, as they\nonly existed to run things where track-upstream was enabled.\n\nChange-Id: I597c1a577c53e2db61413d9ec531378667691d2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ef03e7fa36e0de5f89abf3bdea10fdd8eea151d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ef03e7fa36e0de5f89abf3bdea10fdd8eea151d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
