)]}'
{"id":"openstack%2Fproject-config~799822","triplet_id":"openstack%2Fproject-config~master~I551c62ecf243b199cdf01ae6823197fd674d0874","project":"openstack/project-config","branch":"master","topic":"osa/vault","hashtags":[],"change_id":"I551c62ecf243b199cdf01ae6823197fd674d0874","subject":"Create repo for Hashicorp Vault deployment","status":"MERGED","created":"2021-07-07 13:18:35.000000000","updated":"2021-07-14 08:23:48.000000000","submitted":"2021-07-14 06:59:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799822-osa/vault","meta_rev_id":"ff7dc3e541855a25f29ebf98255817d795df08ab","_number":799822,"virtual_id_number":799822,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-14 06:59:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-07-13 16:41:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2021-07-14 06:42:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2021-07-14 06:42:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-07 13:30:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-13 16:41:45.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-07-14 06:42:17.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"a2c726cf193f14dd852974dac57e129c205bff07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-07 13:18:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66ae248b49af85c227a3d284589f2a17b07d4665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 13:30:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a43cd98aba444a82a3559e2b47cc0b8d : SUCCESS in 6m 16s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0192e5f276bc44199cb72fa254bd08bf : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"adee9bca53a43709888c413eeb90837afb75c7aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-09 08:36:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"46bcc6ed75a31d943c226ade76fe26b8da7bac29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 08:49:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fa8b6951dd994727923a130d6e32511f : SUCCESS in 10m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7e3bf31b915e4608a747f4e750e44a0f : SUCCESS in 4m 18s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/50d9f45fe7a342b8a2d9440d950d5e84 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":2},{"id":"21bba866fd595cd8368292c6ddc4c09fc7bbc5a0","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-07-13 16:41:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6014f80bc1a270650bd4a31bb173e333f9126b09","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-14 06:42:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0abf9e2d8847e277fcb4b04bf228cd883c73a0c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 06:42:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"88ac8ba03063aeec21e6d27960652c632d0e51da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 06:59:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/910e7fa4b61d4dab8830948cb79d5e8e : SUCCESS in 7m 09s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ad4439562cb94c5fa377799700e0addc : SUCCESS in 5m 20s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/d7412ff936454f48a08fa93bba81e881 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":2},{"id":"83c5b05ad825ac1f2c20652a369d71a26596d2b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 06:59:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ff7dc3e541855a25f29ebf98255817d795df08ab","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 08:23:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/adae9c125c524c92af85e3a64cd2b85a : SUCCESS in 3m 11s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/6a4ae57644744f99ae43cd7ef248d7a9 : SUCCESS in 7m 33s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/f9b03d1c6bd34cf79983530b4f2034b4 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9a665e587d210f34c4c4d76819b3ca97f72d5302","revisions":{"dbdd89451d115242bc3de7b64b917b3ce2126481":{"kind":"REWORK","_number":1,"created":"2021-07-07 13:18:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/799822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/799822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/799822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/799822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/799822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/799822/1"}}},"commit":{"parents":[{"commit":"6b78c7d5e3d7f3b1e42a05cf194c9d705a65c5ea","subject":"afsdocs_secret-openstack-manuals: Zuul 4.6.0 fix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6b78c7d5e3d7f3b1e42a05cf194c9d705a65c5ea"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-07-07 13:18:08.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-07-07 13:18:08.000000000","tz":180},"subject":"Create repo for Hashicorp Vault deployment","message":"Create repo for Hashicorp Vault deployment\n\nAs part of implementation of the spec [1] we aim to offer basic Vault\ndeployment out of the box as a tooling required for providing such\nkind of feature.\n\n[1] https://specs.openstack.org/openstack/openstack-ansible-specs/specs/xena/protecting-plaintext-configs.html\n\nChange-Id: I551c62ecf243b199cdf01ae6823197fd674d0874\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dbdd89451d115242bc3de7b64b917b3ce2126481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dbdd89451d115242bc3de7b64b917b3ce2126481"}]},"branch":"refs/heads/master"},"9a665e587d210f34c4c4d76819b3ca97f72d5302":{"kind":"REWORK","_number":2,"created":"2021-07-09 08:36:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/799822/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/799822/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/799822/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/799822/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/799822/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/799822/2"}}},"commit":{"parents":[{"commit":"6b78c7d5e3d7f3b1e42a05cf194c9d705a65c5ea","subject":"afsdocs_secret-openstack-manuals: Zuul 4.6.0 fix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6b78c7d5e3d7f3b1e42a05cf194c9d705a65c5ea"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-07-07 13:18:08.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-07-09 08:36:49.000000000","tz":0},"subject":"Create repo for Hashicorp Vault deployment","message":"Create repo for Hashicorp Vault deployment\n\nAs part of implementation of the spec [1] we aim to offer basic Vault\ndeployment out of the box as a tooling required for providing such\nkind of feature.\n\n[1] https://specs.openstack.org/openstack/openstack-ansible-specs/specs/xena/protecting-plaintext-configs.html\n\nChange-Id: I551c62ecf243b199cdf01ae6823197fd674d0874\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9a665e587d210f34c4c4d76819b3ca97f72d5302"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9a665e587d210f34c4c4d76819b3ca97f72d5302"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
