)]}'
{"id":"openstack%2Fproject-config~799824","triplet_id":"openstack%2Fproject-config~master~I814b91e55dd0b1ccf12e6c1c2a780d06c6b6ff05","project":"openstack/project-config","branch":"master","topic":"osa/vault","hashtags":[],"change_id":"I814b91e55dd0b1ccf12e6c1c2a780d06c6b6ff05","subject":"Add Vault role to zuul config","status":"ABANDONED","created":"2021-07-07 13:22:07.000000000","updated":"2021-07-09 08:35:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"23e1ed4d9930466f1d866a6161b20d6907baa211","_number":799824,"virtual_id_number":799824,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-07 13:29:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2021-07-07 15:23:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-07 13:29:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-07 15:23:27.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"d5f49a7e63d866182d81e0f26abc9a66109e61d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-07 13:22:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e15ce8542c52b31111b20e75b2e1598b0d1073a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 13:29:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8cbd2e0c874647c495d62e8a25e71ef2 : SUCCESS in 6m 24s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6e8368faba7446c1b2d250f84f577794 : SUCCESS in 5m 03s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/3122e221a89a46a8b53a88d7d963fd8f : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7280a1ba63a8fd226f9478d8b7e9fba89807416d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-07 15:23:27.000000000","message":"Patch Set 1: Code-Review-1\n\nMerge it with https://review.opendev.org/c/openstack/project-config/+/799822/1 - no need for separate submits.","accounts_in_message":[],"_revision_number":1},{"id":"23e1ed4d9930466f1d866a6161b20d6907baa211","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-09 08:35:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff2c65b92b6c54f905948573e3fe180c6558cdb4","revisions":{"ff2c65b92b6c54f905948573e3fe180c6558cdb4":{"kind":"REWORK","_number":1,"created":"2021-07-07 13:22:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/799824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/799824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/799824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/799824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/799824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/799824/1"}}},"commit":{"parents":[{"commit":"dbdd89451d115242bc3de7b64b917b3ce2126481","subject":"Create repo for Hashicorp Vault deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dbdd89451d115242bc3de7b64b917b3ce2126481"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-07-07 13:21:16.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-07-07 13:21:16.000000000","tz":180},"subject":"Add Vault role to zuul config","message":"Add Vault role to zuul config\n\nOnce repo is created, we can add it to the zuul projects and gerritbot\n\nChange-Id: I814b91e55dd0b1ccf12e6c1c2a780d06c6b6ff05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ff2c65b92b6c54f905948573e3fe180c6558cdb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ff2c65b92b6c54f905948573e3fe180c6558cdb4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
