)]}'
{"id":"openstack%2Fproject-config~800157","triplet_id":"openstack%2Fproject-config~master~I3528c591dd283cac7ca7a5463f5a24f2c113a9c2","project":"openstack/project-config","branch":"master","topic":"retire-tempest-tripleo","hashtags":[],"change_id":"I3528c591dd283cac7ca7a5463f5a24f2c113a9c2","subject":"tripleo-common-tempest-plugin - Step 4: Remove Project","status":"MERGED","created":"2021-07-09 05:11:11.000000000","updated":"2021-07-25 20:00:09.000000000","submitted":"2021-07-25 19:11:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800157-retire-tempest-tripleo","meta_rev_id":"a8842148a41a99110b46aeeb304a1c213a6d9819","_number":800157,"virtual_id_number":800157,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-25 19:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2021-07-25 18:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-16 14:02:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2021-07-23 18:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2021-07-25 18:59:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-09 05:19:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-12 08:03:56.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-07-12 17:15:45.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-07-23 18:19:11.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"5fa202cdaf6a2cba5b5545c43e3357d9a1e3658c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-09 05:11:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4351e4fd9aa423cdd3368e48d6c76ace34820b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 05:19:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/04ffd5223cea49bd9af7f11a64d75bc2 : SUCCESS in 6m 25s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/15101450842c4d4ba8e9f57740486716 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"33882a3f3bc8307ab3000175505adfc5931f768c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-12 07:53:01.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"00c78801c7a23702e4d458d914a816d0d0cef2ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 08:03:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e558d6fc42b0484f8b0cf3d4f499e3e4 : SUCCESS in 5m 49s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2d9a7a1937a24db08dd1570bc68fa130 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5aeeb2647ea435a447405dd125ad0037744cd1a4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-12 08:03:56.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9272ad23160d02cf6cc409c8e8e465e77815a28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-12 08:11:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9a6a08c6a5ea728bcfb0745121a828049bb81a97","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-12 08:11:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7b287eb9315b926fe12cfb5d23f12b46ed56fd13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 08:18:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b3bc21d401df4f10b0dd113833cb6b5a : SUCCESS in 5m 23s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e5ddd94954ab485b862ad08df326b01a : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":3},{"id":"f42304bf4158e12b06f408e5e64bba3ad6f7309b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-12 08:32:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"41490dd3ba9cbf3a1f8c6d593a5f43352729955b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-12 17:15:45.000000000","message":"Patch Set 3: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":3},{"id":"8d21ed594d2fa197333a7059e932abb863ab4b31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-16 13:58:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f927339253b83580e4137471c7e9e648e3ccebca","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-16 14:02:02.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"78d1922971d54c993c6781091db6e55af54f3755","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 14:17:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/40373d6acc8c4b8fb7736e2de5e4b4af : SUCCESS in 7m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/20b486b87c864e9f91c9bc391061cd28 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":4},{"id":"ce5a145e528b6fc870c230627d74afe5041f04a3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-07-23 18:19:11.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM but the dependency has not yet merged.","accounts_in_message":[],"_revision_number":4},{"id":"e0ad349212d2ffaff5a2d6e8ecfd88b77e1590bf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-25 18:59:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"415ce6079d324cba8c815e686f0e08b7bba8719b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-25 19:00:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"86d3c3bbc4f05b17ba7f6821d30cba9417ddb7d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-25 19:11:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/86242739542d4e3da2bfc3b5ec98b122 : SUCCESS in 5m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ce9b9503b85a473b8c390afe00dbe9d1 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":4},{"id":"18c29ebb12e572a22a49eafca6cd002f484a45bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-25 19:11:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"a8842148a41a99110b46aeeb304a1c213a6d9819","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-25 20:00:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/bef9dcfb0eca47ff9597b0ff42afc4a0 : SUCCESS in 2m 39s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/5276cc61b5a74d799a809d777137b7f8 : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"57bb2df7b32927dbe9bc74c463cddfdd4d98a264","revisions":{"cb12d71ab75551216299481c15924b5d9560fd4f":{"kind":"REWORK","_number":1,"created":"2021-07-09 05:11:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/800157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/800157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/800157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/800157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/800157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/800157/1"}}},"commit":{"parents":[{"commit":"a1150ae523af15579f57ee9f754780651ce9b0ac","subject":"Retire tripleo-common-tempest-plugin - Step 1: End project Gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a1150ae523af15579f57ee9f754780651ce9b0ac"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-09 05:10:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-09 05:10:35.000000000","tz":540},"subject":"Retire tripleo-common-tempest-plugin - Step 3: Remove Project","message":"Retire tripleo-common-tempest-plugin - Step 3: Remove Project\n\nDepends-on: https://review.opendev.org/800155\nChange-Id: I3528c591dd283cac7ca7a5463f5a24f2c113a9c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb12d71ab75551216299481c15924b5d9560fd4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb12d71ab75551216299481c15924b5d9560fd4f"}]},"branch":"refs/heads/master"},"551f84fa5dc294224e4aae915f09e469381e8ed3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-12 07:53:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/800157/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/800157/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/800157/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/800157/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/800157/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/800157/2"}}},"commit":{"parents":[{"commit":"bd6a81acba2d9f7e9ac107df25c8133dd6350c84","subject":"tripleo-common-tempest-plugin - Step 2: End project Gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bd6a81acba2d9f7e9ac107df25c8133dd6350c84"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-09 05:10:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-12 07:52:25.000000000","tz":540},"subject":"tripleo-common-tempest-plugin - Step 4: Remove Project","message":"tripleo-common-tempest-plugin - Step 4: Remove Project\n\nDepends-on: https://review.opendev.org/c/openstack/tripleo-common-tempest-plugin/+/800155\nChange-Id: I3528c591dd283cac7ca7a5463f5a24f2c113a9c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/551f84fa5dc294224e4aae915f09e469381e8ed3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/551f84fa5dc294224e4aae915f09e469381e8ed3"}]},"branch":"refs/heads/master"},"2bc167c426ecdd197f0db8223663c18087f5533d":{"kind":"REWORK","_number":3,"created":"2021-07-12 08:11:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/800157/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/800157/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/800157/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/800157/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/800157/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/800157/3"}}},"commit":{"parents":[{"commit":"bd6a81acba2d9f7e9ac107df25c8133dd6350c84","subject":"tripleo-common-tempest-plugin - Step 2: End project Gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bd6a81acba2d9f7e9ac107df25c8133dd6350c84"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-09 05:10:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-12 08:11:23.000000000","tz":540},"subject":"tripleo-common-tempest-plugin - Step 4: Remove Project","message":"tripleo-common-tempest-plugin - Step 4: Remove Project\n\nDepends-on: https://review.opendev.org/c/openstack/tripleo-common-tempest-plugin/+/800155\nChange-Id: I3528c591dd283cac7ca7a5463f5a24f2c113a9c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2bc167c426ecdd197f0db8223663c18087f5533d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2bc167c426ecdd197f0db8223663c18087f5533d"}]},"branch":"refs/heads/master"},"57bb2df7b32927dbe9bc74c463cddfdd4d98a264":{"kind":"REWORK","_number":4,"created":"2021-07-16 13:58:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/800157/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/800157/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/800157/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/800157/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/800157/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/800157/4"}}},"commit":{"parents":[{"commit":"1add2274953cf82a45e495c74fb1ead691d41586","subject":"tripleo-common-tempest-plugin - Step 2: End project Gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1add2274953cf82a45e495c74fb1ead691d41586"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-09 05:10:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 13:58:04.000000000","tz":540},"subject":"tripleo-common-tempest-plugin - Step 4: Remove Project","message":"tripleo-common-tempest-plugin - Step 4: Remove Project\n\nDepends-on: https://review.opendev.org/c/openstack/tripleo-common-tempest-plugin/+/800155\nChange-Id: I3528c591dd283cac7ca7a5463f5a24f2c113a9c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57bb2df7b32927dbe9bc74c463cddfdd4d98a264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57bb2df7b32927dbe9bc74c463cddfdd4d98a264"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
