)]}'
{"id":"openstack%2Fproject-config~800196","triplet_id":"openstack%2Fproject-config~master~I228bffa92bfea6c88ba4a1ef6e0306b64f9b4831","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I228bffa92bfea6c88ba4a1ef6e0306b64f9b4831","subject":"register andd return back tricircle under x namespaces","status":"ABANDONED","created":"2021-07-09 10:08:37.000000000","updated":"2021-07-12 07:52:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"238baadd7448672f6bafd641b0b73508db010bd8","_number":800196,"virtual_id_number":800196,"owner":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-09 10:17:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-09 10:17:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c1cb239cc5b8f87d44d7d743a981712a1db22fd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"date":"2021-07-09 10:08:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b75e1f6d5d452cb9adbcf4f05b3c89473897009f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 10:17:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0b4b6479d6c34ae2ba6d31a496d0b054 : SUCCESS in 6m 23s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f5f45dea13c24abab777f23a378a15d2 : FAILURE in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"238baadd7448672f6bafd641b0b73508db010bd8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"date":"2021-07-12 07:52:57.000000000","message":"Abandoned\n\nnew commit 800427","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d10662fb01b7f5dffc0003887b7b93d1e4b914d1","revisions":{"d10662fb01b7f5dffc0003887b7b93d1e4b914d1":{"kind":"REWORK","_number":1,"created":"2021-07-09 10:08:37.000000000","uploader":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"ref":"refs/changes/96/800196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/800196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/800196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/800196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/800196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/800196/1"}}},"commit":{"parents":[{"commit":"cc37470cd803f0ec2e706c83c999c6e67d5849cb","subject":"Merge \"update channel config with new registered nic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cc37470cd803f0ec2e706c83c999c6e67d5849cb"}]}],"author":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2021-07-09 10:07:59.000000000","tz":0},"committer":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2021-07-09 10:07:59.000000000","tz":0},"subject":"register andd return back tricircle under x namespaces","message":"register andd return back tricircle under x namespaces\n\nChange-Id: I228bffa92bfea6c88ba4a1ef6e0306b64f9b4831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d10662fb01b7f5dffc0003887b7b93d1e4b914d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d10662fb01b7f5dffc0003887b7b93d1e4b914d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
