)]}'
{"id":"openstack%2Fproject-config~800241","triplet_id":"openstack%2Fproject-config~master~I1460b5e1989e28cf0c355c16c2b9a5cfb92ed8bd","project":"openstack/project-config","branch":"master","topic":"panko_leftover","hashtags":[],"change_id":"I1460b5e1989e28cf0c355c16c2b9a5cfb92ed8bd","subject":"Enable deleting left-overs of panko deprecation","status":"ABANDONED","created":"2021-07-09 13:34:50.000000000","updated":"2021-07-22 15:02:04.000000000","total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"938ead0c551c00dd1863572c977603572cd3b2a6","_number":800241,"virtual_id_number":800241,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-16 10:06:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2021-07-22 13:31:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":-1,"date":"2021-07-21 18:43:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-09 13:42:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-11 09:01:33.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-07-21 18:43:12.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"6eeb317a270b82e0f5b52a099cd964f075c6a1f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-09 13:34:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb253fad80373413a0b83c04ade7108de82fb815","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-09 13:35:29.000000000","message":"Patch Set 1:\n\nThis patch is to remove the rest of panko https://review.opendev.org/c/openstack/project-config/+/800241","accounts_in_message":[],"_revision_number":1},{"id":"342699b159b72bc77852e225ffdcb06ccbce8848","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-09 13:37:48.000000000","message":"Patch Set 1:\n\nThis one indeed: https://review.opendev.org/c/openstack/panko/+/800167","accounts_in_message":[],"_revision_number":1},{"id":"27ea5360d0ed5d1738b564a7353b42d6b6d7e316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 13:42:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/13ea9e579056419eaa3c3d84de7b6d2f : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"76784ca7523c214e416bd36fea3b45ca7da34c06","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-11 09:01:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0b7c7c44ae9c9bffda8f4c1c1408e96e5575adf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-16 09:56:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba84ba7c7c48ae1935b8f72091e8c120dc7f6da3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 10:06:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5b69714a37f548f5bc4ef94443bfd08c : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":2},{"id":"c4eec16283142e0c69c0b24c1296671dacaf4780","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-21 18:43:12.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff1ac7f500386acf78e8fc525be4526db79c6386","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-22 12:57:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"940eb0bea6396170ca6336498121f7f6b5633e03","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-22 13:31:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9f7cd37538c3e20962ea7ba2a4f0232e6b7fe45b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-22 14:15:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d8ee508e4e04a57320dc5c819f58c42125a0dca1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-22 15:01:22.000000000","message":"Abandoned\n\nNot necessary anymore","accounts_in_message":[],"_revision_number":2},{"id":"938ead0c551c00dd1863572c977603572cd3b2a6","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-22 15:02:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"02460adbb31667754003bf265eb72bd22e4391dd","revisions":{"b4a0774648f5c7047a2a3b80872c8da9914522df":{"kind":"REWORK","_number":1,"created":"2021-07-09 13:34:50.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/41/800241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/800241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/800241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/800241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/800241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/800241/1"}}},"commit":{"parents":[{"commit":"cc37470cd803f0ec2e706c83c999c6e67d5849cb","subject":"Merge \"update channel config with new registered nic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cc37470cd803f0ec2e706c83c999c6e67d5849cb"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-09 13:33:19.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-09 13:33:19.000000000","tz":120},"subject":"Enable deleting left-overs of panko deprecation","message":"Enable deleting left-overs of panko deprecation\n\nUnfortunately, there were some left-overs during\ncontent removal in the panko repository.\nThis is to get rid of that.\n\nChange-Id: I1460b5e1989e28cf0c355c16c2b9a5cfb92ed8bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b4a0774648f5c7047a2a3b80872c8da9914522df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b4a0774648f5c7047a2a3b80872c8da9914522df"}]},"branch":"refs/heads/master"},"02460adbb31667754003bf265eb72bd22e4391dd":{"kind":"REWORK","_number":2,"created":"2021-07-16 09:56:26.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/41/800241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/800241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/800241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/800241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/800241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/800241/2"}}},"commit":{"parents":[{"commit":"6ab94b627518f8116297e2ac95338f6ac3b7305c","subject":"nodepool: set inmotion cloud to zero","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6ab94b627518f8116297e2ac95338f6ac3b7305c"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-09 13:33:19.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-16 09:56:14.000000000","tz":120},"subject":"Enable deleting left-overs of panko deprecation","message":"Enable deleting left-overs of panko deprecation\n\nUnfortunately, there were some left-overs during\ncontent removal in the panko repository.\nThis is to get rid of that.\n\nChange-Id: I1460b5e1989e28cf0c355c16c2b9a5cfb92ed8bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/02460adbb31667754003bf265eb72bd22e4391dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/02460adbb31667754003bf265eb72bd22e4391dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
