)]}'
{"id":"openstack%2Fproject-config~800427","triplet_id":"openstack%2Fproject-config~master~I4f3271d3a8e9f78043c7e43ac3d52b44aa2ae683","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I4f3271d3a8e9f78043c7e43ac3d52b44aa2ae683","subject":"register and bring back tricircle under x namespaces","status":"ABANDONED","created":"2021-07-12 07:39:35.000000000","updated":"2021-07-13 10:43:21.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9a418b056c2bf72292c094bf2f85c58962377b02","_number":800427,"virtual_id_number":800427,"owner":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-12 07:49:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2021-07-12 08:07:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-12 07:49:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-12 08:07:36.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"e58c04f9295c9faba827fa454daf990eb18a175c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"date":"2021-07-12 07:39:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6161fae5da520d848bb5def364a3f08fe20b95c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 07:49:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6162423256964390a20fe5ff0d6b96d0 : SUCCESS in 5m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/08304846e1354a15966df9867c75e428 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"667d64697fda013c4a9055b05f600f39d31461c2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-12 08:07:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nPlease read:\nhttps://docs.opendev.org/opendev/infra-manual/latest/creators.html\n\nYou need to update zuul/main.yaml as well and I expect you want to import to old git repository with its history.","accounts_in_message":[],"_revision_number":1},{"id":"9a418b056c2bf72292c094bf2f85c58962377b02","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-13 10:43:21.000000000","message":"Abandoned\n\nDuplicate of https://review.opendev.org/c/openstack/project-config/+/800442","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d57e3efa72bb512839e52446c7ff2ab6a30d31e8","revisions":{"d57e3efa72bb512839e52446c7ff2ab6a30d31e8":{"kind":"REWORK","_number":1,"created":"2021-07-12 07:39:35.000000000","uploader":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"ref":"refs/changes/27/800427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/27/800427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/800427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/800427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/800427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/27/800427/1"}}},"commit":{"parents":[{"commit":"5719cb656ad043c2475f035c5a9e9d68659880c8","subject":"Add channel ops to #openstack-manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5719cb656ad043c2475f035c5a9e9d68659880c8"}]}],"author":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2021-07-12 07:38:32.000000000","tz":0},"committer":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2021-07-12 07:38:32.000000000","tz":0},"subject":"register and bring back tricircle under x namespaces","message":"register and bring back tricircle under x namespaces\n\nChange-Id: I4f3271d3a8e9f78043c7e43ac3d52b44aa2ae683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d57e3efa72bb512839e52446c7ff2ab6a30d31e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d57e3efa72bb512839e52446c7ff2ab6a30d31e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
