)]}'
{"id":"openstack%2Fproject-config~800459","triplet_id":"openstack%2Fproject-config~master~Ifa691c2d2e7ac8bb502c8fae7dbdcc89fa4ef825","project":"openstack/project-config","branch":"master","topic":"retire-django-openstack-auth","hashtags":[],"change_id":"Ifa691c2d2e7ac8bb502c8fae7dbdcc89fa4ef825","subject":"Retire django-openstack-auth","status":"MERGED","created":"2021-07-12 12:08:42.000000000","updated":"2021-07-21 20:14:09.000000000","submitted":"2021-07-21 18:53:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"800459-retire-django-openstack-auth","meta_rev_id":"13e3a19ad3299283d1c7ca3dde43b1931e2266d0","_number":800459,"virtual_id_number":800459,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-21 18:53:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2021-07-19 05:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2021-07-21 20:14:09.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2021-07-16 16:47:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2021-07-21 18:41:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2021-07-21 18:41:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}],"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-12 12:20:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-12 15:15:23.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-07-12 17:24:22.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-07-12 17:48:17.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2021-07-15 16:03:31.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-07-21 18:41:23.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"}],"messages":[{"id":"5a08098d7bb57b233f3a33a1c87e8bac96e575a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-12 12:08:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35721968b731db26864b64209bc0c6bbaec73248","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-12 12:09:18.000000000","message":"Topic set to retire-django-openstack-auth","accounts_in_message":[],"_revision_number":1},{"id":"d0b83a49e739e18941d28d0b1eb6213c6b663406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 12:20:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d9734431c3d5403f9a0724b3da7ab725 : SUCCESS in 5m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/763ae322086b44e4a07fc4a7c9f15e8a : SUCCESS in 5m 26s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/926ec7581fbd4031bc52db1d51bcf798 : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c9a31d639bb129c72c584d643216fed686d78c94","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-12 15:15:23.000000000","message":"Patch Set 1: Code-Review-1\n\nBefore doing this, django-openstack-auth needs to be dropped from openstack/requirements stable/pike.\nhttps://docs.openstack.org/project-team-guide/repository.html#step-1-stop-requirements-syncing-if-set-up","accounts_in_message":[],"_revision_number":1},{"id":"c055b180b87656c03fbc357d21e26ab98b8ebd89","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-12 15:26:58.000000000","message":"Patch Set 1:\n\nCould you explain why the retirement process can be started from Step 4 of the process in the project team guide? I know why but it is worth describing it in the commit message.\n\nI see your patches corresponding to step 4 (in the project team guide; step 3 in the opendev manual) and step 6, but step 5 looks needed as https://docs.openstack.org/django_openstack_auth/latest/ exists.\n\nIn addition, I personally would like to make stable/pike EOL before retiring the django-openstack-auth repo, so it would be nice if https://review.opendev.org/c/openstack/releases/+/800493 lands before this so that \"pike-eol\" tag can be pushed to the github mirror of the openstack repositories.","accounts_in_message":[],"_revision_number":1},{"id":"c6d87ad1f0ddea6d8b3dd66c1450325caaa5e86c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-12 17:24:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nalong with amotoki comment, one more thing to take care.","accounts_in_message":[],"_revision_number":1},{"id":"15f56de47e44aa7a4ca958bf94fe0483f8261a94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-12 17:37:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"925dd4ce0c957833ca522e68ee0c76203c97feb3","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-07-12 17:48:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5fdfb75426b88a12b593fc935df161e56caeb67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 17:49:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1676f7a65b164ba183d7a2ac40a8d0a4 : SUCCESS in 5m 37s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3184283eb07a45349cb2addbb337f01c : SUCCESS in 4m 26s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/063b6d23306144d19581d98d3dcbffdc : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":2},{"id":"09799794a4a5a26f6a51298d3b5c4e1fcf2ce951","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-12 18:02:53.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Could you explain why the retirement process can be started from Step 4 of the process in the project team guide? I know why but it is worth describing it in the commit message.\n\u003e \n\u003e\u003eDone.\n\u003e I see your patches corresponding to step 4 (in the project team guide; step 3 in the opendev manual) and step 6, but step 5 looks needed as https://docs.openstack.org/django_openstack_auth/latest/ exists.\n\u003e \n\u003e\u003e pushed a patch for step 5 https://review.opendev.org/c/openstack/openstack-manuals/+/800522 .\n\u003e In addition, I personally would like to make stable/pike EOL before retiring the django-openstack-auth repo, so it would be nice if https://review.opendev.org/c/openstack/releases/+/800493 lands before this so that \"pike-eol\" tag can be pushed to the github mirror of the openstack repositories.\n\u003e\u003e agreed, +1.","accounts_in_message":[],"_revision_number":2},{"id":"37f4113bfe33df4857a703fa6fe8805e7514d783","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-12 18:04:03.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bb756e36294e6f5c81bc2e1517c3c26528859d07","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-13 23:29:34.000000000","message":"Patch Set 2: Code-Review+1\n\nlgtm thanks","accounts_in_message":[],"_revision_number":2},{"id":"cd7c5f29f0e8ed39b8010b1a1a637e604dab0ab6","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-14 04:32:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nIn addition to my inline comment, could you mark this as -W until pike-eol is tagged?\nPerhaps there is no strong dependency between this change and pike-eol (+ requirements stable/pike change) but it is better to make pike-eol (and requirement change) happen before the repo retirement as the horizon team should be responsible for them. After the repo retirement, TC will be resposinble for the repo.","accounts_in_message":[],"_revision_number":2},{"id":"8b67a761f015890d017dca2ab2549096da84a019","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-14 06:25:55.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84a50c5a3f3c71e239b1fe4d4f2dbb02f92caded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-14 18:21:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"87a74ab5af04e61b1946c912e92d344bd4da2dc2","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-14 18:27:09.000000000","message":"Patch Set 3:\n\nI will remove ``zuul/main.yaml`` reference in follow-up patch\nhttps://review.opendev.org/c/openstack/project-config/+/800532","accounts_in_message":[],"_revision_number":3},{"id":"50646bb406416ab4726f64d4e853e5d7bd38366c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 18:32:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/73c04c6d59334be984ebd53571ae2c2c : SUCCESS in 6m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2f20bec14bfe40bb9360fc231cfe3fb6 : SUCCESS in 4m 36s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/7a29a7cebf1d4d208fbfc5ccad17b9c0 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":3},{"id":"06fcfc1e03e6bc298b087e2fbe237943fbf32fa7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-15 16:03:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6954b7b797c6c78597726b979e449073cf3e2ccf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-15 16:05:52.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nPlease also update the other change on top of this.","accounts_in_message":[],"_revision_number":3},{"id":"ef5b7a50dfd6a0126876a44607acfb000b71dc14","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-15 16:27:54.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"21132914d4273a09cccb7292b1ebcdf9718a8c32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-15 19:37:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8c22e4762cc6c2edc795e7827d8c8d08a7a69d33","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-15 19:38:15.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"12c980bdb3345493c3117fd668000c4497dae4de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 19:49:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a16d24a60b374c13a4da990267cff103 : SUCCESS in 6m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1425be1fb33a41e6989148be09ade575 : SUCCESS in 4m 48s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/286cabc872e54ec79eccb499df628530 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":4},{"id":"f1166c8350a023376e4b052a6c3ef36f6a55e1fb","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-16 02:57:23.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"43ca76f0d8bb99324b69f52181afebf9f5790ee3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-16 07:25:10.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"172cb01156a30c3fe47846b14131384408c93c85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-16 08:07:13.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"45a52267a9e173fe4000fb4f9fce5cb530443c07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 08:16:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3777175545704a0ab393e7abea063117 : SUCCESS in 6m 31s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0cebeaee4f654d3ea7e51d4eb9a523fc : SUCCESS in 6m 14s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/36c343690b0149059091a009608fc21b : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":5},{"id":"6aaf55b541196c00141ef0be26957e3fa4007d1e","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-16 10:13:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6b1afbbbf8de89d62e4065be86a0f8d0ae950754","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-16 16:43:48.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"da86b8adb9ed0b332e493129cb010b9340f3303f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-16 16:47:09.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"5dfa18ada41d680f31c97325baa0579d7a454cda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 16:55:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e4ed9bbace1848dba2c33240ca05aac9 : SUCCESS in 6m 09s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2126013a31f6473aaa9565dbbb787034 : SUCCESS in 4m 41s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/a8d6f0eae148486c84a293bf233c3f81 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":6},{"id":"f7ba11df72d2f1fcfafb9c145ad392e968697687","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-19 05:17:04.000000000","message":"Patch Set 6:\n\nHi Andreas, is there anything left that I should update for the retirement process?","accounts_in_message":[],"_revision_number":6},{"id":"b047ad97d0be5c2ae4f2ce391af850c87d8d48d6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-19 05:46:30.000000000","message":"Patch Set 6: Code-Review+2\n\nall good now","accounts_in_message":[],"_revision_number":6},{"id":"a3353e72f9c0805aa0f5513c973e441102669201","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-07-21 18:41:23.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"f05610d743a94e4b0c10fd03af2dcac6722d3577","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 18:42:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1c306238af3c26a7ed5f82af89d769adebeee177","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 18:53:59.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5372cf3a08fd449484ac7c8b284e6148 : SUCCESS in 6m 10s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3f78080067f241eeb3e819185f8841d8 : SUCCESS in 4m 36s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/6a31d3828e13451092586e26fab4b3f2 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":6},{"id":"df913da0f14a8ec8918008f701a8330b3e134e3a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 18:53:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"ebe3d26127338c011600a5f607cb2f92ff896bd8","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 20:14:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/42501af250ba4163bc64e547e2856825 : SUCCESS in 3m 07s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/de41471467a242bfb49d8032f2abd963 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"0cb4819d91ddf35165cd8f439fd801bbf58badac","revisions":{"bf49fcdc4da388d965e1635c9ecdc2c996e44ef5":{"kind":"REWORK","_number":1,"created":"2021-07-12 12:08:42.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/59/800459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/800459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/800459/1"}}},"commit":{"parents":[{"commit":"5719cb656ad043c2475f035c5a9e9d68659880c8","subject":"Add channel ops to #openstack-manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5719cb656ad043c2475f035c5a9e9d68659880c8"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-12 12:06:04.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-12 12:08:38.000000000","tz":330},"subject":"Retire django-openstack-auth","message":"Retire django-openstack-auth\n\n``django_openstack_auth`` code was merged into the main horizon repository\nduring the queens release. This repo is already deprecated and now we can\nretire it. This patch removes all references of ``django_openstack_auth``\nas mentioned in [1].\n\n[1] https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nChange-Id: Ifa691c2d2e7ac8bb502c8fae7dbdcc89fa4ef825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bf49fcdc4da388d965e1635c9ecdc2c996e44ef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bf49fcdc4da388d965e1635c9ecdc2c996e44ef5"}]},"branch":"refs/heads/master"},"1efb161216ad0dad30696f9f6caee81de6af4b07":{"kind":"REWORK","_number":2,"created":"2021-07-12 17:37:52.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/59/800459/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/800459/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/800459/2"}}},"commit":{"parents":[{"commit":"5719cb656ad043c2475f035c5a9e9d68659880c8","subject":"Add channel ops to #openstack-manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5719cb656ad043c2475f035c5a9e9d68659880c8"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-12 12:06:04.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-12 17:35:08.000000000","tz":330},"subject":"Retire django-openstack-auth","message":"Retire django-openstack-auth\n\n``django_openstack_auth`` code was merged into the main horizon\nrepository during the queens release. This repo is already\ndeprecated and now we can retire it. This patch removes all\nreferences of ``django_openstack_auth`` as mentioned in [1].\nAlso, this patch takes care of step 2(End project Gating)\nand and step 4(Remove Project) because\nstep 1(Stop requirements syncing) and step3(Remove Project Content)\nis  already handled by If04778ccc99ea92355378d59e616d8794e36ea14\nand I74b10a90fe79fc768cfb8de6f68d3cd2f4938e51 respectively.\n\n[1] https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nChange-Id: Ifa691c2d2e7ac8bb502c8fae7dbdcc89fa4ef825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1efb161216ad0dad30696f9f6caee81de6af4b07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1efb161216ad0dad30696f9f6caee81de6af4b07"}]},"branch":"refs/heads/master"},"49ead63bb7132cf6741207245c5459bcee05cf1b":{"kind":"REWORK","_number":3,"created":"2021-07-14 18:21:35.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/59/800459/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/800459/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/800459/3"}}},"commit":{"parents":[{"commit":"5719cb656ad043c2475f035c5a9e9d68659880c8","subject":"Add channel ops to #openstack-manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5719cb656ad043c2475f035c5a9e9d68659880c8"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-12 12:06:04.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-14 18:19:38.000000000","tz":330},"subject":"Retire django-openstack-auth","message":"Retire django-openstack-auth\n\n``django_openstack_auth`` code was merged into the main horizon\nrepository during the queens release. This repo is already\ndeprecated and now we can retire it. This patch removes all\nreferences of ``django_openstack_auth`` as mentioned in [1].\nAlso, this patch takes care of step 2(End project Gating)\nand and step 4(Remove Project) because\nstep 1(Stop requirements syncing) and step3(Remove Project Content)\nis  already handled by If04778ccc99ea92355378d59e616d8794e36ea14\nand I74b10a90fe79fc768cfb8de6f68d3cd2f4938e51 respectively.\n\n[1] https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nChange-Id: Ifa691c2d2e7ac8bb502c8fae7dbdcc89fa4ef825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/49ead63bb7132cf6741207245c5459bcee05cf1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/49ead63bb7132cf6741207245c5459bcee05cf1b"}]},"branch":"refs/heads/master"},"5244954ae2e34de2903c974ff87498ddcfecca6d":{"kind":"REWORK","_number":4,"created":"2021-07-15 19:37:40.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/59/800459/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/800459/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/800459/4"}}},"commit":{"parents":[{"commit":"5719cb656ad043c2475f035c5a9e9d68659880c8","subject":"Add channel ops to #openstack-manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5719cb656ad043c2475f035c5a9e9d68659880c8"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-12 12:06:04.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-15 19:31:22.000000000","tz":330},"subject":"Retire django-openstack-auth","message":"Retire django-openstack-auth\n\n``django_openstack_auth`` code was merged into the main horizon\nrepository during the queens release. This repo is already\ndeprecated and now we can retire it. This patch removes all\nreferences of ``django_openstack_auth`` as mentioned in [1].\nAlso, this patch takes care of step 2(End project Gating)\nand and step 4(Remove Project) because\nstep 1(Stop requirements syncing) and step3(Remove Project Content)\nis  already handled by If04778ccc99ea92355378d59e616d8794e36ea14\nand I74b10a90fe79fc768cfb8de6f68d3cd2f4938e51 respectively.\n\nNote: this patch doesn\u0027t drop the official-openstack-repo-jobs template\nand noop-job because noop-job is needs to be defined for the check and\ngate queues to merge a patch and official-openstack-repo-jobs is needed\nto syncup the changed in github repo once Ib811fb321d18fc01f3786f8b3ab16b2eda558864\nmerged we can drop this template and noop-job.\n\n[1] https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nChange-Id: Ifa691c2d2e7ac8bb502c8fae7dbdcc89fa4ef825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5244954ae2e34de2903c974ff87498ddcfecca6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5244954ae2e34de2903c974ff87498ddcfecca6d"}]},"branch":"refs/heads/master"},"223002cf8611115a6b4f881c526c9c3642d7f568":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-07-16 08:07:13.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/59/800459/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/800459/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/800459/5"}}},"commit":{"parents":[{"commit":"01d16dad9920be0a2958ae2975e0a95a0b461182","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/01d16dad9920be0a2958ae2975e0a95a0b461182"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-12 12:06:04.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-16 07:54:54.000000000","tz":330},"subject":"Retire django-openstack-auth","message":"Retire django-openstack-auth\n\n``django_openstack_auth`` code was merged into the main horizon\nrepository during the queens release. This repo is already\ndeprecated and now we can retire it. This patch removes all\nreferences of ``django_openstack_auth`` as mentioned in [1].\nAlso, this patch takes care of step 2(End project Gating)\nand and step 4(Remove Project) because\nstep 1(Stop requirements syncing) and step3(Remove Project Content)\nis  already handled by If04778ccc99ea92355378d59e616d8794e36ea14\nand I74b10a90fe79fc768cfb8de6f68d3cd2f4938e51 respectively.\n\nNote: this patch doesn\u0027t drop the official-openstack-repo-jobs template\nand noop-job because noop-job is needs to be defined for the check and\ngate queues to merge a patch and official-openstack-repo-jobs is needed\nto syncup the changed in github repo once Ib811fb321d18fc01f3786f8b3ab16b2eda558864\nmerged we can drop this template and noop-job.\n\n[1] https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nChange-Id: Ifa691c2d2e7ac8bb502c8fae7dbdcc89fa4ef825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/223002cf8611115a6b4f881c526c9c3642d7f568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/223002cf8611115a6b4f881c526c9c3642d7f568"}]},"branch":"refs/heads/master"},"0cb4819d91ddf35165cd8f439fd801bbf58badac":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-07-16 16:43:48.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/59/800459/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/800459/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/800459/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/800459/6"}}},"commit":{"parents":[{"commit":"6ab94b627518f8116297e2ac95338f6ac3b7305c","subject":"nodepool: set inmotion cloud to zero","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6ab94b627518f8116297e2ac95338f6ac3b7305c"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-12 12:06:04.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-16 16:40:16.000000000","tz":330},"subject":"Retire django-openstack-auth","message":"Retire django-openstack-auth\n\n``django_openstack_auth`` code was merged into the main horizon\nrepository during the queens release. This repo is already\ndeprecated and now we can retire it. This patch removes all\nreferences of ``django_openstack_auth`` as mentioned in [1].\nAlso, this patch takes care of step 2(End project Gating)\nand and step 4(Remove Project) because\nstep 1(Stop requirements syncing) and step3(Remove Project Content)\nis  already handled by If04778ccc99ea92355378d59e616d8794e36ea14\nand I74b10a90fe79fc768cfb8de6f68d3cd2f4938e51 respectively.\n\nNote: this patch doesn\u0027t drop the official-openstack-repo-jobs template\nand noop-job because noop-job is needs to be defined for the check and\ngate queues to merge a patch and official-openstack-repo-jobs is needed\nto syncup the changed in github repo once Ib811fb321d18fc01f3786f8b3ab16b2eda558864\nmerged we can drop this template and noop-job.\n\n[1] https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nChange-Id: Ifa691c2d2e7ac8bb502c8fae7dbdcc89fa4ef825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0cb4819d91ddf35165cd8f439fd801bbf58badac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0cb4819d91ddf35165cd8f439fd801bbf58badac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
