)]}'
{"id":"openstack%2Fproject-config~800532","triplet_id":"openstack%2Fproject-config~master~Iaab900d48aaa2f21abb048403138d03628ad7617","project":"openstack/project-config","branch":"master","topic":"retire-django-openstack-auth","hashtags":[],"change_id":"Iaab900d48aaa2f21abb048403138d03628ad7617","subject":"Retire django-openstack-auth","status":"MERGED","created":"2021-07-12 18:00:59.000000000","updated":"2021-07-23 19:01:41.000000000","submitted":"2021-07-23 18:27:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"800532-retire-django-openstack-auth","meta_rev_id":"e2b6e6e4119e632a43644f5e69ab38b26ec18b50","_number":800532,"virtual_id_number":800532,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-23 18:27:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2021-07-19 05:44:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2021-07-21 19:45:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2021-07-16 16:47:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-23 18:19:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-23 18:19:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-12 18:07:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-12 18:21:48.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2021-07-14 06:45:00.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-07-16 16:47:21.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-07-21 19:45:06.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-07-22 18:38:44.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2021-07-23 18:19:58.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"144789f9b36ebc568ccdb3df2cdff09a5164cdde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-12 18:00:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3818bf59fdaac1023146d7bb4a46cc94a2f8f86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 18:07:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f7eaeaf511ef408b9c0cdc0da900cb7e : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"31cc2a44e8ce128b2b5fa1f874857bbeb8bb453e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-12 18:10:38.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ce4b95f6ba5af850add00ca723d3c3c5415e1e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 18:18:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0123bb61f2a64185a634650c4de32501 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":2},{"id":"663749d2659e7c965217b1b3c7eae8a3beb55994","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-12 18:21:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4c62f3df0d6b245681d2ad43d8a600d9ef5068b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-12 19:08:41.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"92b23b4798076109b63df43a63e29f1783997c9f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-12 19:09:21.000000000","message":"Topic set to etire-django-openstack-auth","accounts_in_message":[],"_revision_number":3},{"id":"eda48a9e7b2400b80576198bec8906ee01c8e509","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-12 19:09:35.000000000","message":"Topic etire-django-openstack-auth removed","accounts_in_message":[],"_revision_number":3},{"id":"fd83068ffe797ff260f54a7399c3d6bd7f8a4ab1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-12 19:09:41.000000000","message":"Topic set to retire-django-openstack-auth","accounts_in_message":[],"_revision_number":3},{"id":"1338e9ca866985ce325cfe8f1634028e9e3d673e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 19:20:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2f51e13f75ae428ba7d58221618a9c3d : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"9eacc50dfda5bdc70d09efb68a20a5f19a669e10","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-14 06:45:00.000000000","message":"Patch Set 3: Code-Review-1\n\nIf you retire it, check the list for retirement and update all places:\n- gerrit/projects.yaml\n- zuul/main.yaml\n- gerritbot/channels.yaml\n- roles/copy-release-tools-scripts/files/release-tools/functions","accounts_in_message":[],"_revision_number":3},{"id":"ab79b0aa14014035c94f7797f27c27e0ac41b8d4","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-14 11:58:21.000000000","message":"Patch Set 3: Workflow-1\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e If you retire it, check the list for retirement and update all places:\n\u003e - gerrit/projects.yaml\n\u003e - zuul/main.yaml\n\u003e - gerritbot/channels.yaml\n\u003e - roles/copy-release-tools-scripts/files/release-tools/functions\n\nHI Andreas, I have already updated above files except ``roles/copy-release-tools-scripts/files/release-tools/functions`` in [1]. I will update remaining file soon\nin [1].\nthanks. \n\n\n[1] https://review.opendev.org/c/openstack/project-config/+/800459","accounts_in_message":[],"_revision_number":3},{"id":"c300606924822009b3c378e1cb2d2880efb7823f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-14 12:58:06.000000000","message":"Patch Set 3:\n\nI see the other now. This is wrong - add zuul/main.yaml to this change as well, otherwise you cannot merge the README change.","accounts_in_message":[],"_revision_number":3},{"id":"31067da6fd36dc320912d71a257f3953240d26f0","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-14 13:35:34.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e I see the other now. This is wrong - add zuul/main.yaml to this change as well, otherwise you cannot merge the README change.\n\nok will do that, what about \"roles/copy-release-tools-scripts/files/release-tools/functions\".\nlooks like I messed up with order:(.","accounts_in_message":[],"_revision_number":3},{"id":"fc26a6b287c4f6f7b4ed246e03f61b8479d2a1e1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-14 14:23:43.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 3:\n\u003e \u003e \n\u003e \u003e I see the other now. This is wrong - add zuul/main.yaml to this change as well, otherwise you cannot merge the README change.\n\u003e \n\u003e ok will do that, what about \"roles/copy-release-tools-scripts/files/release-tools/functions\".\n\u003e looks like I messed up with order:(.\n\nDoesn\u0027t matter where you add the functions change.","accounts_in_message":[],"_revision_number":3},{"id":"921a55b68fc0822b24805beea2e0eff9b6e0ea3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-16 16:43:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f91200a0fd6e1fc0185983cf77c04fdf3096a567","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-16 16:47:21.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c34fba2cc4fcd2f2ad92cce60410af92c65b5db2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 16:56:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6a1e5041f39a4f6ebc6524ebd0d54089 : SUCCESS in 6m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6bcec8cafe9344af8bb5e9586cbbb1af : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":4},{"id":"1ae2dfb32878283e00c04c51566627fab5493724","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-19 05:44:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b26ea601adfcb630b1b291b682a62431a909038b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-21 19:45:06.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ac9ca76a735998a227eee1b20030037463e46b7a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-07-23 18:19:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2487822e80ad7001a2358c3e00efaeb625bd3cb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 18:20:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"152efe1297b42f46de027aaadcf2005eace6299f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 18:27:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7d56935a3ac84195bc4889f60d156f94 : SUCCESS in 5m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e2cee5670cd14d089e38e3ae07282e40 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":4},{"id":"2495620ecf91351dde70f5e3db41f4891cbd3722","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 18:27:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"e2b6e6e4119e632a43644f5e69ab38b26ec18b50","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 19:01:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/8928fccc114f41169b07320951ea564f : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"30368c6e55b206f109bb50046bfcff73a4a49621","revisions":{"de215aee375d94577ac3829d17ba8ceb3619c10a":{"kind":"REWORK","_number":1,"created":"2021-07-12 18:00:59.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/32/800532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/800532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/800532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/800532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/800532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/800532/1"}}},"commit":{"parents":[{"commit":"1efb161216ad0dad30696f9f6caee81de6af4b07","subject":"Retire django-openstack-auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1efb161216ad0dad30696f9f6caee81de6af4b07"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-12 17:58:53.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-12 17:58:53.000000000","tz":330},"subject":"Retire django-openstack-auth","message":"Retire django-openstack-auth\n\nThis patch is a follow-up patch for\nIfa691c2d2e7ac8bb502c8fae7dbdcc89fa4ef825nwhich removes\n\"django_openstack_auth\" project template from ``zuul.d/projects.yaml``.\n\nChange-Id: Iaab900d48aaa2f21abb048403138d03628ad7617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/de215aee375d94577ac3829d17ba8ceb3619c10a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/de215aee375d94577ac3829d17ba8ceb3619c10a"}]},"branch":"refs/heads/master"},"049f97ee048da3053041e8703250d9ee834eb3d4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-12 18:10:38.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/32/800532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/800532/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/800532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/800532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/800532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/800532/2"}}},"commit":{"parents":[{"commit":"1efb161216ad0dad30696f9f6caee81de6af4b07","subject":"Retire django-openstack-auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1efb161216ad0dad30696f9f6caee81de6af4b07"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-12 17:58:53.000000000","tz":330},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2021-07-12 18:10:28.000000000","tz":0},"subject":"Retire django-openstack-auth","message":"Retire django-openstack-auth\n\nThis patch is a follow-up patch for\nIfa691c2d2e7ac8bb502c8fae7dbdcc89fa4ef825 which removes\n\"django_openstack_auth\" project template reference from ``zuul.d/projects.yaml``.\n\nChange-Id: Iaab900d48aaa2f21abb048403138d03628ad7617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/049f97ee048da3053041e8703250d9ee834eb3d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/049f97ee048da3053041e8703250d9ee834eb3d4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f8c64286d50553f38af128630bde1320acd29a31":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-07-12 19:08:41.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/32/800532/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/800532/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/800532/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/800532/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/800532/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/800532/3"}}},"commit":{"parents":[{"commit":"1efb161216ad0dad30696f9f6caee81de6af4b07","subject":"Retire django-openstack-auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1efb161216ad0dad30696f9f6caee81de6af4b07"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-12 17:58:53.000000000","tz":330},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2021-07-12 19:08:34.000000000","tz":0},"subject":"Retire django-openstack-auth","message":"Retire django-openstack-auth\n\nThis patch is a follow-up patch for\nIfa691c2d2e7ac8bb502c8fae7dbdcc89fa4ef825 which removes\n\"django_openstack_auth\" project template reference from ``zuul.d/projects.yaml``.\n\nDepends-On: https://review.opendev.org/c/openstack/django_openstack_auth/+/800133\n\nChange-Id: Iaab900d48aaa2f21abb048403138d03628ad7617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f8c64286d50553f38af128630bde1320acd29a31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f8c64286d50553f38af128630bde1320acd29a31"}]},"branch":"refs/heads/master","description":"Edit commit message"},"30368c6e55b206f109bb50046bfcff73a4a49621":{"kind":"REWORK","_number":4,"created":"2021-07-16 16:43:48.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/32/800532/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/800532/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/800532/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/800532/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/800532/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/800532/4"}}},"commit":{"parents":[{"commit":"0cb4819d91ddf35165cd8f439fd801bbf58badac","subject":"Retire django-openstack-auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0cb4819d91ddf35165cd8f439fd801bbf58badac"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-12 17:58:53.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-16 16:43:23.000000000","tz":330},"subject":"Retire django-openstack-auth","message":"Retire django-openstack-auth\n\nThis patch is a follow-up patch for\nIfa691c2d2e7ac8bb502c8fae7dbdcc89fa4ef825 which removes\n\"django_openstack_auth\" project template reference from ``zuul.d/projects.yaml``.\n\nDepends-On: https://review.opendev.org/c/openstack/django_openstack_auth/+/800133\n\nChange-Id: Iaab900d48aaa2f21abb048403138d03628ad7617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/30368c6e55b206f109bb50046bfcff73a4a49621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/30368c6e55b206f109bb50046bfcff73a4a49621"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
