)]}'
{"id":"openstack%2Fproject-config~800885","triplet_id":"openstack%2Fproject-config~master~I0e2a9d453e2e8c0036c9094220855a4701bff7f3","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I0e2a9d453e2e8c0036c9094220855a4701bff7f3","subject":"register and bring back tricircle under x namespaces","status":"ABANDONED","created":"2021-07-15 07:00:26.000000000","updated":"2021-07-15 07:11:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5788171aca05f21418ad5c37e7d4da64b47c7c29","_number":800885,"virtual_id_number":800885,"owner":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-15 07:07:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-15 07:07:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40cffd1d97ac1aed0999e50ac5286e28af5d1bf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"date":"2021-07-15 07:00:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a9352894155de4a7fc2f6eb3a0460dc94e8986f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 07:07:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d67069d106704e719532f781dcbeb98d : SUCCESS in 5m 37s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6d7f41ba3d1346fd8ac23a6c03f94fb1 : FAILURE in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5788171aca05f21418ad5c37e7d4da64b47c7c29","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"date":"2021-07-15 07:11:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b14769076adee0e11dc4e5e6304c6afcb535a90f","revisions":{"b14769076adee0e11dc4e5e6304c6afcb535a90f":{"kind":"REWORK","_number":1,"created":"2021-07-15 07:00:26.000000000","uploader":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"ref":"refs/changes/85/800885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/85/800885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/800885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/800885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/800885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/85/800885/1"}}},"commit":{"parents":[{"commit":"110c736d1de60a22ea06217e1da48db77f3bc0fb","subject":"Merge \"Remove publish-to-pypi from retired neutron-lbaas repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/110c736d1de60a22ea06217e1da48db77f3bc0fb"}]}],"author":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2021-07-15 06:59:57.000000000","tz":0},"committer":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2021-07-15 06:59:57.000000000","tz":0},"subject":"register and bring back tricircle under x namespaces","message":"register and bring back tricircle under x namespaces\n\nChange-Id: I0e2a9d453e2e8c0036c9094220855a4701bff7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b14769076adee0e11dc4e5e6304c6afcb535a90f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b14769076adee0e11dc4e5e6304c6afcb535a90f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
