)]}'
{"id":"openstack%2Fproject-config~801376","triplet_id":"openstack%2Fproject-config~master~I9fbbed3131e2ebfd590f27c569603add15bc969d","project":"openstack/project-config","branch":"master","topic":"fix-released-stable-branch","hashtags":[],"change_id":"I9fbbed3131e2ebfd590f27c569603add15bc969d","subject":"Set launchpad bug Fix Released after adding comment","status":"MERGED","created":"2021-07-19 18:54:24.000000000","updated":"2021-10-07 16:38:58.000000000","submitted":"2021-10-07 16:38:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"801376-fix-released-stable-branch","meta_rev_id":"26ef60fc76551795f72867c847066706f7e73aad","_number":801376,"virtual_id_number":801376,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-07 16:38:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2021-10-07 09:18:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-09-03 11:58:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-02 23:06:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-07 16:30:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-19 19:03:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-21 19:39:51.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-07-21 22:30:36.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-09-02 08:16:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-02 16:55:22.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-09-02 16:57:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-09-02 23:06:52.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-09-09 08:44:47.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2021-10-07 09:18:58.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"b39c633aa3514ecdddaed70cd754414981b9df51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-19 18:54:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"925893ddf4f530c2051118121eabffe6512fde1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 19:03:45.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/921268faf00048f19bc289caca245d2b : FAILURE in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8a3ac72357eb309268791d8245562bce04ceb614","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-19 19:33:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"85fffd636a276dbcedab4fc70f8b69a694a662e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 19:43:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6948155f79554907a952385a058074d9 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"fe40950dddcdb55e787a11f341e459ed658f52be","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-21 19:39:51.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nGenerally looks good. Minor suggestions inline.","accounts_in_message":[],"_revision_number":2},{"id":"e8d6b1740dfbf3d5bdddafee5679b029dea73f61","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-21 20:55:07.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bb7dcb194315b52e598f5b501e9e78a7cc6ca65a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-21 21:00:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"149be5bf664db2a33ff2d6c7cb70282f7ddcb687","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-21 21:01:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f5cd2c2eb6494c603730598bb6d1b61df8b2b6ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-21 21:01:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"af5b6e4a43f9419c1ae4ebc7ce2c40704681adff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 21:08:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3c24db5241a0447d8d37f3eac65ff4e8 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":4},{"id":"f0bccd89ce836324e1547325240ef373dc899b61","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-22 10:24:29.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)\n\nThanks melwitt! I did not know that this is the right place to add this feature.\n\n(I used a simimlar script [1] (with launchpad in read-only mode as a first approach) that printed direct links and in that way it was really easy to set the bugs as released, but of course this is waaaay more better as this is fully automated. Thanks! :))\n\n[1] https://review.opendev.org/c/openstack/nova/+/774223","accounts_in_message":[],"_revision_number":4},{"id":"0d3d8d6b7cf9776b82c01c3583006e9e7c6d9e5d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-22 16:53:47.000000000","message":"Patch Set 4:\n\n(2 comments)\n\n\u003e Patch Set 4: Code-Review+1\n\u003e \n\u003e (2 comments)\n\u003e \n\u003e Thanks melwitt! I did not know that this is the right place to add this feature.\n\u003e \n\u003e (I used a simimlar script [1] (with launchpad in read-only mode as a first approach) that printed direct links and in that way it was really easy to set the bugs as released, but of course this is waaaay more better as this is fully automated. Thanks! :))\n\u003e \n\u003e [1] https://review.opendev.org/c/openstack/nova/+/774223\n\nAh, I wondered what kind of script you were using, thanks for showing that!","accounts_in_message":[],"_revision_number":4},{"id":"7d7a7617e74636a77aa37091c8df545cd21c7bd8","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-22 22:17:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"30fd4f2350bbc6b977a2bde39bf8eceeeb9c3a57","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-25 22:57:22.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2c4f073bccb2feaf1e3a1e4b3c86f4f99f283b2b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-02 08:16:01.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9c05eede22276319f4fa71f592e2ab0693859c7a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-02 16:55:22.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)\n\n+1 for the general idea; I guess it would work fine without amending but it can be improved per comments.","accounts_in_message":[],"_revision_number":4},{"id":"93312ebddc185744ba752804e43ced6bc916496d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-02 22:25:29.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"25885cbc71553c50a28ac91b01ab8065d6cf20b1","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-02 22:43:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c5cbc4af19904905d36d6a0f9a2170d39a1f52f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-02 22:45:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"543e3474c5edec4cc09832ba5ac932f5a64c1962","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 22:57:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4bbe0b15503f495888724e7a032f05c5 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":5},{"id":"abd6abca52da63b66e777a9e947a1ec7365e8ed9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-09-02 23:06:52.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"3c448dbd8c533d88308e60bd3492d5164d3358a9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-03 11:58:15.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ccd58541dd8c82aa9875a04718fbf666961f40de","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-03 11:59:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3d4b6d9b1183d987bd0be2e31f7106c4d06bcbc1","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-09-09 08:44:47.000000000","message":"Patch Set 5:\n\nI\u0027m a bit worried about changing that in the last weeks of the cycle -- this may affect how fast we can process final releases (Launchpad API being notoriously slow and this doubling the number of calls we make). I\u0027d feel better if we could sample it before we run it on release day. Let\u0027s discuss it on release meeting","accounts_in_message":[],"_revision_number":5},{"id":"3155ab365567efb1cd481b3681b79ae7a7431cdf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-09 10:46:05.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e I\u0027m a bit worried about changing that in the last weeks of the cycle -- this may affect how fast we can process final releases (Launchpad API being notoriously slow and this doubling the number of calls we make). I\u0027d feel better if we could sample it before we run it on release day. Let\u0027s discuss it on release meeting\n\nYeah, I agree to test it during cooler times. 😉","accounts_in_message":[],"_revision_number":5},{"id":"dffd7ed12ab2946cbe61e2be5320c0771fb82a94","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-10 07:20:31.000000000","message":"Patch Set 5:\n\n+1 to wait a bit and to let pass the rush.","accounts_in_message":[],"_revision_number":5},{"id":"c3e57303e79e425c90b051efde77119b89642837","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-10-07 09:18:58.000000000","message":"Patch Set 5: Code-Review+1\n\nThis is probably fine now. Just ping on #openstack-release when you approve it so that we can watch for any impact.","accounts_in_message":[],"_revision_number":5},{"id":"f541246e868439e72d504af66d0f4dbd3dba9237","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-07 09:28:11.000000000","message":"Patch Set 5:\n\nyeah, let\u0027s merge","accounts_in_message":[],"_revision_number":5},{"id":"8216c1de6f2ebe5a2c69e48fbb7e0ab4e0ad5c12","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-10-07 16:30:10.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"28d69e167eab065d994a4ef5fea1caa9429b9ae7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 16:30:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e42ac41c693cc7b6b4c47b26cd7a4cb91648c90a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 16:38:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ec50b3c564c84cb5a1285222ce442812 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":5},{"id":"26ef60fc76551795f72867c847066706f7e73aad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 16:38:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b6bcacee655b93f4777c7ac8270a889de7cb4aea","revisions":{"1bddf8cbdd028f040bb2411cfd7f76091866695e":{"kind":"REWORK","_number":1,"created":"2021-07-19 18:54:24.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/76/801376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/801376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/801376/1"}}},"commit":{"parents":[{"commit":"143922e7103866e4457a35255b569e3667f98e93","subject":"Merge \"Correct branch path in afsdocs_secret-releasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/143922e7103866e4457a35255b569e3667f98e93"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-07-19 18:49:50.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-07-19 18:49:50.000000000","tz":0},"subject":"Set launchpad bug Fix Released after adding comment","message":"Set launchpad bug Fix Released after adding comment\n\nThis adds a step in the launchpad interaction during a release to set\nthe bug task status to \"Fix Released\" for the series being released.\n\nChange-Id: I9fbbed3131e2ebfd590f27c569603add15bc969d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1bddf8cbdd028f040bb2411cfd7f76091866695e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1bddf8cbdd028f040bb2411cfd7f76091866695e"}]},"branch":"refs/heads/master"},"13787b90893bb636f3960f6c88fb6311bbb88684":{"kind":"REWORK","_number":2,"created":"2021-07-19 19:33:06.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/76/801376/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/801376/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/801376/2"}}},"commit":{"parents":[{"commit":"143922e7103866e4457a35255b569e3667f98e93","subject":"Merge \"Correct branch path in afsdocs_secret-releasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/143922e7103866e4457a35255b569e3667f98e93"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-07-19 18:49:50.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-07-19 19:29:29.000000000","tz":0},"subject":"Set launchpad bug Fix Released after adding comment","message":"Set launchpad bug Fix Released after adding comment\n\nThis adds a step in the launchpad interaction during a release to set\nthe bug task status to \"Fix Released\" for the series being released.\n\nChange-Id: I9fbbed3131e2ebfd590f27c569603add15bc969d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/13787b90893bb636f3960f6c88fb6311bbb88684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/13787b90893bb636f3960f6c88fb6311bbb88684"}]},"branch":"refs/heads/master"},"b9a38287b9f398163f667fb6349d1ec50978f559":{"kind":"REWORK","_number":3,"created":"2021-07-21 21:00:16.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/76/801376/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/801376/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/801376/3"}}},"commit":{"parents":[{"commit":"143922e7103866e4457a35255b569e3667f98e93","subject":"Merge \"Correct branch path in afsdocs_secret-releasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/143922e7103866e4457a35255b569e3667f98e93"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-07-19 18:49:50.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-07-21 20:59:46.000000000","tz":0},"subject":"Set launchpad bug Fix Released after adding comment","message":"Set launchpad bug Fix Released after adding comment\n\nThis adds a step in the launchpad interaction during a release to set\nthe bug task status to \"Fix Released\" for the series being released.\n\nChange-Id: I9fbbed3131e2ebfd590f27c569603add15bc969d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b9a38287b9f398163f667fb6349d1ec50978f559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b9a38287b9f398163f667fb6349d1ec50978f559"}]},"branch":"refs/heads/master"},"aca81aa6fac5039a640a1ee6d1d6f27ad378a876":{"kind":"REWORK","_number":4,"created":"2021-07-21 21:01:47.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/76/801376/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/801376/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/801376/4"}}},"commit":{"parents":[{"commit":"143922e7103866e4457a35255b569e3667f98e93","subject":"Merge \"Correct branch path in afsdocs_secret-releasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/143922e7103866e4457a35255b569e3667f98e93"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-07-19 18:49:50.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-07-21 21:01:35.000000000","tz":0},"subject":"Set launchpad bug Fix Released after adding comment","message":"Set launchpad bug Fix Released after adding comment\n\nThis adds a step in the launchpad interaction during a release to set\nthe bug task status to \"Fix Released\" for the series being released.\n\nChange-Id: I9fbbed3131e2ebfd590f27c569603add15bc969d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aca81aa6fac5039a640a1ee6d1d6f27ad378a876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aca81aa6fac5039a640a1ee6d1d6f27ad378a876"}]},"branch":"refs/heads/master"},"b6bcacee655b93f4777c7ac8270a889de7cb4aea":{"kind":"REWORK","_number":5,"created":"2021-09-02 22:45:42.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/76/801376/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/801376/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/801376/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/801376/5"}}},"commit":{"parents":[{"commit":"143922e7103866e4457a35255b569e3667f98e93","subject":"Merge \"Correct branch path in afsdocs_secret-releasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/143922e7103866e4457a35255b569e3667f98e93"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-07-19 18:49:50.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-09-02 22:45:31.000000000","tz":0},"subject":"Set launchpad bug Fix Released after adding comment","message":"Set launchpad bug Fix Released after adding comment\n\nThis adds a step in the launchpad interaction during a release to set\nthe bug task status to \"Fix Released\" for the series being released.\n\nChange-Id: I9fbbed3131e2ebfd590f27c569603add15bc969d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b6bcacee655b93f4777c7ac8270a889de7cb4aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b6bcacee655b93f4777c7ac8270a889de7cb4aea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
