)]}'
{"id":"openstack%2Fproject-config~803411","triplet_id":"openstack%2Fproject-config~master~Id71888115597894d4dcb5c12731f819177fd9341","project":"openstack/project-config","branch":"master","topic":"centos-8-stream-wheel","hashtags":[],"change_id":"Id71888115597894d4dcb5c12731f819177fd9341","subject":"Add CentOS 8 Stream wheel publish jobs","status":"MERGED","created":"2021-08-04 02:06:21.000000000","updated":"2021-08-10 02:55:04.000000000","submitted":"2021-08-10 02:36:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803411-centos-8-stream-wheel","meta_rev_id":"ff022057a3deb77d05f64bae666038c9df20dea8","_number":803411,"virtual_id_number":803411,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 02:36:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-08-05 22:59:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2021-08-06 09:13:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-10 02:28:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-04 02:18:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-05 22:59:30.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-08-06 09:13:56.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"620a7edaf14a98ec0ed20f0f31888d55b556572f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-08-04 02:06:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb33357d27ecae5468cdeaedbbda78d1a1b9191a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-08-04 02:07:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"994328ac6379fd33e7ea77451853470e852de44a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 02:18:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7f1f259328394639b7531fa943090040 : FAILURE in 5m 26s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/bf0122d72aa74548b7b67439e7562530 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"08bb421c2396add799dd560964d8a7c190dd4254","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-08-04 03:33:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"90160c90a9546c2e0caf99ff92a6e7e00e467f32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 03:40:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/acd092c7409a4b179cd614fd9446585c : FAILURE in 5m 21s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/0d1d844ff3614c61abd88c735e539b20 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":3},{"id":"cf989b5723bbcaa0c5dbfd46c1e3c45227788fad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-08-04 22:20:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c01ba58e67db71ab68f69b48b9005e6371aeb720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 22:20:39.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Unknown projects: x/tap-as-a-service\n\nThe error appears in the following project stanza:\n\n  project:\n      name: x/vmware-nsx\n      templates:\n        - periodic-jobs-with-neutron-lib-master\n        - publish-to-pypi\n      check:\n        queue: vmware-nsx\n      gate:\n        queue: vmware-nsx\n      periodic:\n        jobs:\n          - openstack-tox-py36-with-neutron-lib-master:\n              timeout: 5400\n              required-projects:\n                - openstack/networking-sfc\n                - openstack/neutron\n                - openstack/neutron-dynamic-routing\n                - openstack/neutron-fwaas\n                - openstack/neutron-vpnaas\n                - openstack/octavia\n                - x/networking-l2gw\n                - x/tap-as-a-service\n                - x/vmware-nsxlib\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 8270, column 3","accounts_in_message":[],"_revision_number":4},{"id":"ebd2a6491f764216f29b383afc6e9f776f9cc4f2","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-08-04 23:21:16.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"98533308a4bdfbd609da5e3c138b9856efcb1c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 23:31:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/42ec82ee56f1437eae3816a7f77a4866 : SUCCESS in 5m 37s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/571c0ba2a5a547629297a05e1144b090 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":4},{"id":"48d36c73c6f10370bbedeb822e03c30d014e5771","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-08-05 22:59:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"92628fa4f53ec6398caff40cc619553a0df1c1ca","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-08-06 09:13:56.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM, please self-approve when you are around to monitor that everything works.","accounts_in_message":[],"_revision_number":4},{"id":"ac094e5676f6375802abb453439ee5fceeb611bf","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-08-10 02:28:35.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3fd5e408d8127d16c8015b7a1c568b713979736b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 02:29:05.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"74c74ed3160af589f0c720d9d3d4d8f66df3e383","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 02:36:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/230e8818550348278dc9430af056cd4f : SUCCESS in 6m 12s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/e5f45bb7cf634a3f8b407061b94dd969 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":4},{"id":"9e98c54037dfdcb48a82a2a040094ab4ae072d8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 02:36:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ff022057a3deb77d05f64bae666038c9df20dea8","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 02:55:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/4fabc35f818146db94f035c9562ebc0b : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3d776f6734ff0c5197471c7a89aa84511a0eff88","revisions":{"58e71adcd2a4dad177d30fa117f5579890148421":{"kind":"REWORK","_number":1,"created":"2021-08-04 02:06:21.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/11/803411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/803411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/803411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/803411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/803411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/803411/1"}}},"commit":{"parents":[{"commit":"533ec6ea0e84475a673f9b2a285554be8dad0625","subject":"Merge \"Rename x/tap-as-a-service to openstack/tap-as-a-service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/533ec6ea0e84475a673f9b2a285554be8dad0625"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-08-04 02:05:36.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-08-04 02:05:36.000000000","tz":600},"subject":"Add CentOS 8 Stream wheel publish jobs","message":"Add CentOS 8 Stream wheel publish jobs\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/802988\nChange-Id: Id71888115597894d4dcb5c12731f819177fd9341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/58e71adcd2a4dad177d30fa117f5579890148421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/58e71adcd2a4dad177d30fa117f5579890148421"}]},"branch":"refs/heads/master"},"fa5b94b2cc79eb7499e650a592acb2f158d1c41c":{"kind":"REWORK","_number":2,"created":"2021-08-04 02:07:03.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/11/803411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/803411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/803411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/803411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/803411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/803411/2"}}},"commit":{"parents":[{"commit":"533ec6ea0e84475a673f9b2a285554be8dad0625","subject":"Merge \"Rename x/tap-as-a-service to openstack/tap-as-a-service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/533ec6ea0e84475a673f9b2a285554be8dad0625"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-08-04 02:05:36.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-08-04 02:06:51.000000000","tz":600},"subject":"Add CentOS 8 Stream wheel publish jobs","message":"Add CentOS 8 Stream wheel publish jobs\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/802988\nChange-Id: Id71888115597894d4dcb5c12731f819177fd9341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fa5b94b2cc79eb7499e650a592acb2f158d1c41c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fa5b94b2cc79eb7499e650a592acb2f158d1c41c"}]},"branch":"refs/heads/master"},"39da2f2996f64bc42cc53b558922ae7c6b25285f":{"kind":"REWORK","_number":3,"created":"2021-08-04 03:33:45.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/11/803411/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/803411/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/803411/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/803411/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/803411/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/803411/3"}}},"commit":{"parents":[{"commit":"533ec6ea0e84475a673f9b2a285554be8dad0625","subject":"Merge \"Rename x/tap-as-a-service to openstack/tap-as-a-service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/533ec6ea0e84475a673f9b2a285554be8dad0625"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-08-04 02:05:36.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-08-04 03:33:32.000000000","tz":600},"subject":"Add CentOS 8 Stream wheel publish jobs","message":"Add CentOS 8 Stream wheel publish jobs\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/802988\nChange-Id: Id71888115597894d4dcb5c12731f819177fd9341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/39da2f2996f64bc42cc53b558922ae7c6b25285f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/39da2f2996f64bc42cc53b558922ae7c6b25285f"}]},"branch":"refs/heads/master"},"3d776f6734ff0c5197471c7a89aa84511a0eff88":{"kind":"REWORK","_number":4,"created":"2021-08-04 22:20:18.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/11/803411/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/803411/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/803411/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/803411/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/803411/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/803411/4"}}},"commit":{"parents":[{"commit":"533ec6ea0e84475a673f9b2a285554be8dad0625","subject":"Merge \"Rename x/tap-as-a-service to openstack/tap-as-a-service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/533ec6ea0e84475a673f9b2a285554be8dad0625"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-08-04 02:05:36.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-08-04 22:20:04.000000000","tz":600},"subject":"Add CentOS 8 Stream wheel publish jobs","message":"Add CentOS 8 Stream wheel publish jobs\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/802988\nChange-Id: Id71888115597894d4dcb5c12731f819177fd9341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d776f6734ff0c5197471c7a89aa84511a0eff88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d776f6734ff0c5197471c7a89aa84511a0eff88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
