)]}'
{"id":"openstack%2Fproject-config~803480","triplet_id":"openstack%2Fproject-config~master~Iee2e37f47fee9ee1fb0016f87069d7cf90a30f5d","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Iee2e37f47fee9ee1fb0016f87069d7cf90a30f5d","subject":"Reintroduce x/tap-as-a-service shared ACL","status":"MERGED","created":"2021-08-04 13:41:22.000000000","updated":"2021-08-04 15:57:32.000000000","submitted":"2021-08-04 15:00:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803480","meta_rev_id":"7f11a300e713d419fdb9130743bc2a5af90c5ac5","_number":803480,"virtual_id_number":803480,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-04 15:00:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-08-04 14:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2021-08-04 14:48:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-04 13:54:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-04 14:48:04.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"d8c6496b1e1423d9f6b0b56f0062fd9f8e890c1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-08-04 13:41:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94b10556f3ec2f1bc7119554afe027b05a813239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 13:54:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b6bd5d4afd6148fcb0480e904e2d1999 : SUCCESS in 5m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/455c07af22a44dd2b606f944326d7457 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"f25cc0972faf5666f8bbe2a062a4f1550f8713b4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-08-04 13:55:49.000000000","message":"Patch Set 1:\n\nPer the referenced change\u0027s commit message, it looks like these two repositories can (and should) be retired.","accounts_in_message":[],"_revision_number":1},{"id":"474c502d00e8d56216fc7d785f236a2a43e867ff","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-08-04 14:48:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"942a11a69ee74d50424dea2120aa9c6b1b7ba174","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 14:48:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37b2beaa951d017ad99ac1c4426a480545cd26c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 15:00:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/328ca35002d4408fbfdb8f4d184410dd : SUCCESS in 7m 17s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d227a521566b4f8e84ddcb4bbf0668f2 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9315d73b913f50b36631abea1700f1923c382094","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 15:00:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7f11a300e713d419fdb9130743bc2a5af90c5ac5","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 15:57:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/60fa6f4994ff4c27a6cafaddc5fe82e5 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"52c8cba54d7e7bc85f612b220935ad110212354a","revisions":{"52c8cba54d7e7bc85f612b220935ad110212354a":{"kind":"REWORK","_number":1,"created":"2021-08-04 13:41:22.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/80/803480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/80/803480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/803480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/803480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/803480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/80/803480/1"}}},"commit":{"parents":[{"commit":"533ec6ea0e84475a673f9b2a285554be8dad0625","subject":"Merge \"Rename x/tap-as-a-service to openstack/tap-as-a-service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/533ec6ea0e84475a673f9b2a285554be8dad0625"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-08-04 13:37:39.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-08-04 13:38:37.000000000","tz":0},"subject":"Reintroduce x/tap-as-a-service shared ACL","message":"Reintroduce x/tap-as-a-service shared ACL\n\nWhen x/tap-as-a-service was renamed to openstack/tap-as-a-service in\nIaa6fc7b0a9dca20ea77637a3890c603837ac7014 its ACL moved into the\nopenstack namespace along with it. However, because these move\nchanges are currently untestable, we did not spot that the ACL was\nalso shared by the x/tap-as-a-service-dashboard and\nx/tap-as-a-service-tempest-plugin projects which were not included\nin the rename work.\n\nFor now, add back the old x/tap-as-a-service Gerrit ACL as it\nappeared prior to the move, so that our consistency checks will stop\nfailing unrelated changes. Separately, we can discuss the fate of\nthese other two projects with their stakeholders, and see what can\nbe done to clean this up a bit.\n\nChange-Id: Iee2e37f47fee9ee1fb0016f87069d7cf90a30f5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/52c8cba54d7e7bc85f612b220935ad110212354a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/52c8cba54d7e7bc85f612b220935ad110212354a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
