)]}'
{"id":"openstack%2Fproject-config~804969","triplet_id":"openstack%2Fproject-config~master~I0da07a7d7677c03714b310600f2dc5c70a589e81","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I0da07a7d7677c03714b310600f2dc5c70a589e81","subject":"bring tricircle under x namespaces","status":"ABANDONED","created":"2021-08-18 09:00:49.000000000","updated":"2021-08-18 09:32:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e04c1d64ded9e31b1642ec1f30cf452b3c6e1168","_number":804969,"virtual_id_number":804969,"owner":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-18 09:11:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-18 09:11:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c392dbb92c8ae55efb2fc50a3357d04d1db07ac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"date":"2021-08-18 09:00:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"608e08d789690370d97fbb6fb2bf2888520a45b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 09:11:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bc9b9a989cd04f9b910362b04ad27c1a : SUCCESS in 6m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/965214d0ff444932ae0ad69a6ef2534c : FAILURE in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e04c1d64ded9e31b1642ec1f30cf452b3c6e1168","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"date":"2021-08-18 09:32:45.000000000","message":"Abandoned\n\nabandon","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c307277ea3b6491d790a8519b63bea54610afa1","revisions":{"2c307277ea3b6491d790a8519b63bea54610afa1":{"kind":"REWORK","_number":1,"created":"2021-08-18 09:00:49.000000000","uploader":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"ref":"refs/changes/69/804969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/804969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/804969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/804969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/804969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/804969/1"}}},"commit":{"parents":[{"commit":"01f845486e4fbcd6bd05052e6337df53461166d6","subject":"Merge \"Adjust secrets for developer.o.o\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/01f845486e4fbcd6bd05052e6337df53461166d6"}]}],"author":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2021-08-18 09:00:13.000000000","tz":0},"committer":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2021-08-18 09:00:13.000000000","tz":0},"subject":"bring tricircle under x namespaces","message":"bring tricircle under x namespaces\n\nChange-Id: I0da07a7d7677c03714b310600f2dc5c70a589e81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c307277ea3b6491d790a8519b63bea54610afa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c307277ea3b6491d790a8519b63bea54610afa1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
