)]}'
{"id":"openstack%2Fproject-config~804970","triplet_id":"openstack%2Fproject-config~master~Ie517cce09aecd221c12780c41f747b89411405f6","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ie517cce09aecd221c12780c41f747b89411405f6","subject":"bring tricircle under x namespaces","status":"ABANDONED","created":"2021-08-18 09:04:06.000000000","updated":"2021-08-18 09:32:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f62628e71339d7b829c10d3079cdec8a3289246c","_number":804970,"virtual_id_number":804970,"owner":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-18 09:15:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-18 09:15:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d22dcd1602f3d1d48162dd4eea1c72a205e9ff96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"date":"2021-08-18 09:04:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77e6554dedc6b7865774c35d184aae2b4a656d39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 09:15:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4fd7818906584de082e83118f1a8051d : SUCCESS in 7m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/195990fcb33d4c4f850216846dfa5db0 : FAILURE in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f62628e71339d7b829c10d3079cdec8a3289246c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"date":"2021-08-18 09:32:24.000000000","message":"Abandoned\n\nabandon","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fcfc9291b4a7324cd4809bb57d96777377ee9732","revisions":{"fcfc9291b4a7324cd4809bb57d96777377ee9732":{"kind":"REWORK","_number":1,"created":"2021-08-18 09:04:06.000000000","uploader":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"ref":"refs/changes/70/804970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/70/804970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/804970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/804970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/804970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/70/804970/1"}}},"commit":{"parents":[{"commit":"2c307277ea3b6491d790a8519b63bea54610afa1","subject":"bring tricircle under x namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c307277ea3b6491d790a8519b63bea54610afa1"}]}],"author":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2021-08-18 09:03:20.000000000","tz":0},"committer":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2021-08-18 09:03:20.000000000","tz":0},"subject":"bring tricircle under x namespaces","message":"bring tricircle under x namespaces\n\nChange-Id: Ie517cce09aecd221c12780c41f747b89411405f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fcfc9291b4a7324cd4809bb57d96777377ee9732"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fcfc9291b4a7324cd4809bb57d96777377ee9732"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
