)]}'
{"id":"openstack%2Fproject-config~804977","triplet_id":"openstack%2Fproject-config~master~I5e246b77ded6f536833732e7c4c24aa7a1646dcb","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I5e246b77ded6f536833732e7c4c24aa7a1646dcb","subject":"bring tricircle under x namespaces","status":"ABANDONED","created":"2021-08-18 09:51:07.000000000","updated":"2022-03-31 09:19:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9c658a59e6b6db2ce3fd751acb42b1cb492ba7ea","_number":804977,"virtual_id_number":804977,"owner":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-18 09:59:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-18 09:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"efe4eee86feef6c5d56225b58180e69f4378f7da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"date":"2021-08-18 09:51:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c1d62b038839e3957f74be5ab466305c203ce59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 09:59:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ae7a31791de7495aba95055ed2fc0019 : SUCCESS in 6m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6244da047d3a472482437bffbc87b411 : FAILURE in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"9c658a59e6b6db2ce3fd751acb42b1cb492ba7ea","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"date":"2022-03-31 09:19:31.000000000","message":"Abandoned\n\nconfilict with 800442","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69333f97087286f922d59a36702815c2c4ae032a","revisions":{"69333f97087286f922d59a36702815c2c4ae032a":{"kind":"REWORK","_number":1,"created":"2021-08-18 09:51:07.000000000","uploader":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"ref":"refs/changes/77/804977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/804977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/804977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/804977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/804977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/804977/1"}}},"commit":{"parents":[{"commit":"01f845486e4fbcd6bd05052e6337df53461166d6","subject":"Merge \"Adjust secrets for developer.o.o\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/01f845486e4fbcd6bd05052e6337df53461166d6"}]}],"author":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2021-08-18 09:50:41.000000000","tz":0},"committer":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2021-08-18 09:50:41.000000000","tz":0},"subject":"bring tricircle under x namespaces","message":"bring tricircle under x namespaces\n\nChange-Id: I5e246b77ded6f536833732e7c4c24aa7a1646dcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/69333f97087286f922d59a36702815c2c4ae032a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/69333f97087286f922d59a36702815c2c4ae032a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
