)]}'
{"id":"openstack%2Fproject-config~805102","triplet_id":"openstack%2Fproject-config~master~I982d45a4739685171784cd710c58b83f1ab87a74","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I982d45a4739685171784cd710c58b83f1ab87a74","subject":"bring tricircle under x namespaces","status":"ABANDONED","created":"2021-08-19 02:25:39.000000000","updated":"2021-12-17 17:49:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"42e096e8769e7cd35bbb737e4cedd46e3d9f3134","_number":805102,"virtual_id_number":805102,"owner":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-19 02:35:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 02:35:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-19 09:18:13.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"CC"}],"messages":[{"id":"e40581e43f80b8391869f23c3d2a000012cc98a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"date":"2021-08-19 02:25:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8b126be634762824e47ac000cb508a9d214e6b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 02:35:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/55b3d6e229d5403588055d4714cdf2fb : SUCCESS in 5m 32s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/daa4265a52f541b786cf1c6e860c5ee3 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"16e71ef85d34aeaede0a840bba0cea021087dff7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-08-19 09:18:13.000000000","message":"Patch Set 1:\n\nPlease continue on the original review: https://review.opendev.org/c/openstack/project-config/+/804977\n\nFor details on the workflow:  Read https://docs.opendev.org/opendev/infra-manual/latest/developers.html#updating-a-change","accounts_in_message":[],"_revision_number":1},{"id":"42e096e8769e7cd35bbb737e4cedd46e3d9f3134","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-12-17 17:49:18.000000000","message":"Abandoned\n\nduplicate","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15ddda47c3f66c2a4c7aa3c6daa530fd89414ee5","revisions":{"15ddda47c3f66c2a4c7aa3c6daa530fd89414ee5":{"kind":"REWORK","_number":1,"created":"2021-08-19 02:25:39.000000000","uploader":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"ref":"refs/changes/02/805102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/02/805102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/805102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/805102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/805102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/02/805102/1"}}},"commit":{"parents":[{"commit":"01f845486e4fbcd6bd05052e6337df53461166d6","subject":"Merge \"Adjust secrets for developer.o.o\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/01f845486e4fbcd6bd05052e6337df53461166d6"}]}],"author":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2021-08-19 02:11:45.000000000","tz":0},"committer":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2021-08-19 02:11:45.000000000","tz":0},"subject":"bring tricircle under x namespaces","message":"bring tricircle under x namespaces\n\nChange-Id: I982d45a4739685171784cd710c58b83f1ab87a74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/15ddda47c3f66c2a4c7aa3c6daa530fd89414ee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/15ddda47c3f66c2a4c7aa3c6daa530fd89414ee5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
