)]}'
{"id":"openstack%2Fproject-config~806788","triplet_id":"openstack%2Fproject-config~master~Iffd253156133e3c9eba6c1942b4b2a551a4344b7","project":"openstack/project-config","branch":"master","topic":"iweb-mtl01","hashtags":[],"change_id":"Iffd253156133e3c9eba6c1942b4b2a551a4344b7","subject":"INAP mtl01 region is now owned by iWeb","status":"MERGED","created":"2021-08-31 18:30:44.000000000","updated":"2021-09-01 15:57:46.000000000","submitted":"2021-09-01 15:44:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806788-iweb-mtl01","meta_rev_id":"8bb8121127154186ee3cc8d37eb1ebd386425d1e","_number":806788,"virtual_id_number":806788,"owner":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-01 15:44:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-08-31 20:31:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-01 15:31:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-01 15:31:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-31 18:33:02.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-08-31 18:34:08.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2021-08-31 18:39:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-31 20:31:07.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"5fd167ebcc512f423785111e56cac4c13c2eb637","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"date":"2021-08-31 18:30:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"387c51410a87c1ca36fe470b737275f3e4e0ce0d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-08-31 18:33:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bbacae82560ee27483e61a5c8e97934c26e58ea9","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-08-31 18:34:08.000000000","message":"Patch Set 1:\n\nI think it might be easier to add iWeb providers in nodepool then set the inap provider resource limits to 0 and let it clean up after itself. If we do this sort of rename in place switch it might get messy.\n\nAlso we\u0027ll need to update https://opendev.org/opendev/system-config/src/branch/master/playbooks/templates/clouds/bridge_all_clouds.yaml.j2 and https://opendev.org/opendev/system-config/src/branch/master/playbooks/templates/clouds/nodepool_clouds.yaml.j2 with the new name info since the cloud: key in the nodepool configs indexes into those files. Again I think we want to do this as an addition, switch over, then remove the old stuff.","accounts_in_message":[],"_revision_number":1},{"id":"5292dc79194272ca7f6a427e905678e7e2328fa1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-08-31 18:34:31.000000000","message":"Patch Set 1: Code-Review-1\n\nActually, we\u0027ll probably need to add a new provider and take the old provider out of service, otherwise we\u0027ll end up abandoning all the nodes we\u0027ve booted there rather than cleaning them up automatically.","accounts_in_message":[],"_revision_number":1},{"id":"0a491fa58e83a7a215251053aa8eb5e010577201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 18:39:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/63a8a6ded74d469e9633e31d44e3b01a : SUCCESS in 6m 43s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/5d0775afdeab4d7186a4a557abe95ec8 : SUCCESS in 5m 19s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/0168f7be249a48bf9f2a6368a4b4c45a : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"dd3c9d1688b47398dd57442eb4087207baa7ba85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"date":"2021-08-31 18:44:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b6191059e27cad567b6cad7aa1eb7fac932813d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 18:56:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/08f6c1d9ca0e4606944de08edd2a6df1 : SUCCESS in 7m 10s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/3388a345232a4a17836397d175053783 : SUCCESS in 5m 28s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/64d4f80688d440cfb7c492174e73327f : FAILURE in 5m 22s","accounts_in_message":[],"_revision_number":2},{"id":"64af97f3ff0f0e9e18568d165c166deca939f913","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"date":"2021-08-31 19:23:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7508606f1942da26c0dfe28e7c803b38033fdce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 19:32:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/df9c76f010924b94982454ea898a3b69 : SUCCESS in 7m 41s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/672dc852bce94f40b703795d99e27d74 : SUCCESS in 6m 24s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/6bebf9a1c94c4639878229d17e735212 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":3},{"id":"0116b0cd873a7efe4abca889b8623a13f81c2696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-08-31 20:30:09.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"134e2463b2058786a6b1b655b33b819e00a06c9d","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-08-31 20:31:07.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"fa103215ad10bf2f62ec3ed860a29ad31e33ce9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 20:41:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e58c1f77c8fa437b926dd563b2bfadcc : SUCCESS in 5m 48s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/060317c2a49842809f3d7e4501014793 : SUCCESS in 4m 11s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ce621d6f4528448ab9ef88734dbe6b41 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":4},{"id":"4d6eee85e3e7dc08047715205bc64a3c95c36abc","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-09-01 15:31:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a45901a4a36c88df8b660c572b811e3c6e358eb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 15:31:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1ec3176e262a25a75b3732419fbf4d0cc46c64c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 15:44:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/794e77b6c2c94f91a316b034c54fb640 : SUCCESS in 7m 05s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/dce3ecb053ae4a2caf65ce1249c4f70a : SUCCESS in 6m 04s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/e022e077f34648988ce4bdff2fe79cdd : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":4},{"id":"71de71adb730df55ec3629c9dc27da6d61a751a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 15:44:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8bb8121127154186ee3cc8d37eb1ebd386425d1e","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 15:57:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/934cf3965c464a47849e7b93553e9520 : SUCCESS in 8m 28s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/1e101a300d044ab897dcaa2c6a06bf30 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"662e981330ca8534cc6df6d6be04ab22b21eb549","revisions":{"f956224ec877117dbb7907f4d09f0db279af6b39":{"kind":"REWORK","_number":1,"created":"2021-08-31 18:30:44.000000000","uploader":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"ref":"refs/changes/88/806788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/88/806788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/806788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/806788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/806788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/88/806788/1"}}},"commit":{"parents":[{"commit":"82d9fcbea9c017eae17f2a050150294845c74dff","subject":"Merge \"Add os-vif-core to stable maintainers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82d9fcbea9c017eae17f2a050150294845c74dff"}]}],"author":{"name":"Mathieu Gagné","email":"mgagne@iweb.com","date":"2021-08-31 18:21:18.000000000","tz":-240},"committer":{"name":"Mathieu Gagné","email":"mgagne@iweb.com","date":"2021-08-31 18:21:18.000000000","tz":-240},"subject":"INAP mtl01 region is now owned by iWeb","message":"INAP mtl01 region is now owned by iWeb\n\nChange-Id: Iffd253156133e3c9eba6c1942b4b2a551a4344b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f956224ec877117dbb7907f4d09f0db279af6b39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f956224ec877117dbb7907f4d09f0db279af6b39"}]},"branch":"refs/heads/master"},"ad037db866ddddcc80a02f521dbdc5f1de283a7f":{"kind":"REWORK","_number":2,"created":"2021-08-31 18:44:52.000000000","uploader":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"ref":"refs/changes/88/806788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/88/806788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/806788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/806788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/806788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/88/806788/2"}}},"commit":{"parents":[{"commit":"82d9fcbea9c017eae17f2a050150294845c74dff","subject":"Merge \"Add os-vif-core to stable maintainers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82d9fcbea9c017eae17f2a050150294845c74dff"}]}],"author":{"name":"Mathieu Gagné","email":"mgagne@iweb.com","date":"2021-08-31 18:40:43.000000000","tz":-240},"committer":{"name":"Mathieu Gagné","email":"mgagne@iweb.com","date":"2021-08-31 18:44:39.000000000","tz":-240},"subject":"INAP mtl01 region is now owned by iWeb","message":"INAP mtl01 region is now owned by iWeb\n\nDisable current Inap mtl01 region and introduce new iWeb mtl01 region.\n\nChange-Id: Iffd253156133e3c9eba6c1942b4b2a551a4344b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad037db866ddddcc80a02f521dbdc5f1de283a7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad037db866ddddcc80a02f521dbdc5f1de283a7f"}]},"branch":"refs/heads/master"},"ed92a2b4a23c0a0f7d9cb1c240bbf09d47d0d69e":{"kind":"REWORK","_number":3,"created":"2021-08-31 19:23:17.000000000","uploader":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"ref":"refs/changes/88/806788/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/88/806788/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/806788/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/806788/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/806788/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/88/806788/3"}}},"commit":{"parents":[{"commit":"82d9fcbea9c017eae17f2a050150294845c74dff","subject":"Merge \"Add os-vif-core to stable maintainers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82d9fcbea9c017eae17f2a050150294845c74dff"}]}],"author":{"name":"Mathieu Gagné","email":"mgagne@iweb.com","date":"2021-08-31 18:40:43.000000000","tz":-240},"committer":{"name":"Mathieu Gagné","email":"mgagne@iweb.com","date":"2021-08-31 19:23:12.000000000","tz":-240},"subject":"INAP mtl01 region is now owned by iWeb","message":"INAP mtl01 region is now owned by iWeb\n\nDisable current Inap mtl01 region and introduce new iWeb mtl01 region.\n\nChange-Id: Iffd253156133e3c9eba6c1942b4b2a551a4344b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ed92a2b4a23c0a0f7d9cb1c240bbf09d47d0d69e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ed92a2b4a23c0a0f7d9cb1c240bbf09d47d0d69e"}]},"branch":"refs/heads/master"},"662e981330ca8534cc6df6d6be04ab22b21eb549":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-08-31 20:30:09.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/88/806788/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/88/806788/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/806788/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/806788/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/806788/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/88/806788/4"}}},"commit":{"parents":[{"commit":"82d9fcbea9c017eae17f2a050150294845c74dff","subject":"Merge \"Add os-vif-core to stable maintainers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82d9fcbea9c017eae17f2a050150294845c74dff"}]}],"author":{"name":"Mathieu Gagné","email":"mgagne@iweb.com","date":"2021-08-31 18:40:43.000000000","tz":-240},"committer":{"name":"Clark Boylan","email":"cboylan@sapwetik.org","date":"2021-08-31 20:30:02.000000000","tz":0},"subject":"INAP mtl01 region is now owned by iWeb","message":"INAP mtl01 region is now owned by iWeb\n\nDisable current Inap mtl01 region and introduce new iWeb mtl01 region.\n\nDepends-On: I1256a2e24df1c79dea06716ae4dfbcfe119c13f8\nChange-Id: Iffd253156133e3c9eba6c1942b4b2a551a4344b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/662e981330ca8534cc6df6d6be04ab22b21eb549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/662e981330ca8534cc6df6d6be04ab22b21eb549"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
