)]}'
{"id":"openstack%2Fproject-config~807206","triplet_id":"openstack%2Fproject-config~master~Ief7d7bc7fbb5b7c0028ead2d507d28700b54ac71","project":"openstack/project-config","branch":"master","topic":"cinder-lvm","hashtags":[],"change_id":"Ief7d7bc7fbb5b7c0028ead2d507d28700b54ac71","subject":"Add the cinder-lvm charm to Openstack charms","status":"MERGED","created":"2021-09-02 19:01:31.000000000","updated":"2021-09-10 01:39:19.000000000","submitted":"2021-09-10 01:16:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"807206-cinder-lvm","meta_rev_id":"1428024ccbbeb568fc8eff9b6082cbfd90697381","_number":807206,"virtual_id_number":807206,"owner":{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 01:16:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-10 00:51:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2021-09-10 00:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2021-09-10 00:56:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-02 19:22:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 16:28:42.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-09-10 00:55:51.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"faa89c212e9a0b12be136cb2f521a31717c1cede","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"},"date":"2021-09-02 19:01:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13bdaaa71bd46f9cdbd2b9877aebfaa1a2890c1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 19:22:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/22bb79ef2e9046afaac639bfa315bb71 : SUCCESS in 6m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/192269a3d964483693963ec4b2103b9e : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"140c252eac03fce390c180b6623f3d59a41d1591","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-03 14:01:13.000000000","message":"Topic cinder-lvm removed","accounts_in_message":[],"_revision_number":1},{"id":"e67509134c84f8e6c75c334f5530134b892b8a59","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-03 14:01:18.000000000","message":"Topic set to project-update","accounts_in_message":[],"_revision_number":1},{"id":"6aeba785ab2877ab31f99237cfbad7b6b535e8a4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-09 16:28:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e9110f5307045fe2ffb9a10829892840213c6e1","author":{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"},"date":"2021-09-09 16:40:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"148ac0fd80d66395bc12d072e16e20ad16130710","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-09 22:31:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b025995bcf06fc614cae3eb436343586a117e4e","author":{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"},"date":"2021-09-09 22:38:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b71a07e7b5f38ef6e0a2ecc1864516baebf4b82e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"},"date":"2021-09-09 22:39:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8cccd574f3a42f15861a43416fb04f8ae5c46335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 22:51:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/81d32855dee54a72952059f37716173a : SUCCESS in 6m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/bfe2fb089099495e86163f26068e7e99 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":2},{"id":"04baf6663f98e18ec6fde964b0b586895120b8cb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-10 00:51:02.000000000","message":"Patch Set 2: Code-Review+1\n\nthanks, lgtm","accounts_in_message":[],"_revision_number":2},{"id":"a5535904f29e7d42bacaf3174c8021e79fbdff53","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-09-10 00:55:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1d2e9e774afe68b88d8707da3abfd506813e474a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-09-10 00:56:24.000000000","message":"Patch Set 2: Workflow+1\n\nTreating Ghanshyam\u0027s +1 as another +2.","accounts_in_message":[],"_revision_number":2},{"id":"d4a9764eabf0909f05ecd641dbe7417d8cbc3f5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 00:56:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"204b755d8045a416d3c9800747717812e9099c33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 01:16:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5002ca476862443c90f82d9dc17b2b5f : SUCCESS in 6m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5f2b42268a2e4dd5bb89eabcae10de24 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"7591a997bb9a3727b28c5fc1fd001ff424f269f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 01:16:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1428024ccbbeb568fc8eff9b6082cbfd90697381","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 01:39:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/3b631ce29c9f4a86a93f566e00366a86 : SUCCESS in 3m 08s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/726fa55f722c4519bd8bc96a4db072ec : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bc7927f5b0047bf3fbf2331105b612788f00f6e7","revisions":{"e741d15b0d476d10344dc8edbc594d418259fe29":{"kind":"REWORK","_number":1,"created":"2021-09-02 19:01:31.000000000","uploader":{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"},"ref":"refs/changes/06/807206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/807206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/807206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/807206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/807206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/807206/1"}}},"commit":{"parents":[{"commit":"662e981330ca8534cc6df6d6be04ab22b21eb549","subject":"INAP mtl01 region is now owned by iWeb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/662e981330ca8534cc6df6d6be04ab22b21eb549"}]}],"author":{"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","date":"2021-09-02 18:37:26.000000000","tz":-180},"committer":{"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","date":"2021-09-02 19:01:28.000000000","tz":-180},"subject":"Add the cinder-lvm charm to Openstack charms","message":"Add the cinder-lvm charm to Openstack charms\n\nThis charm implements an LVM backend for cinder, acting as\na subordinate charm.\n\nRequired-By: Ib2c481faa02a289c24c945378682e28b40fa6f33\nChange-Id: Ief7d7bc7fbb5b7c0028ead2d507d28700b54ac71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e741d15b0d476d10344dc8edbc594d418259fe29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e741d15b0d476d10344dc8edbc594d418259fe29"}]},"branch":"refs/heads/master"},"bc7927f5b0047bf3fbf2331105b612788f00f6e7":{"kind":"REWORK","_number":2,"created":"2021-09-09 22:39:33.000000000","uploader":{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"},"ref":"refs/changes/06/807206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/807206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/807206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/807206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/807206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/807206/2"}}},"commit":{"parents":[{"commit":"662e981330ca8534cc6df6d6be04ab22b21eb549","subject":"INAP mtl01 region is now owned by iWeb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/662e981330ca8534cc6df6d6be04ab22b21eb549"}]}],"author":{"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","date":"2021-09-02 18:37:26.000000000","tz":-180},"committer":{"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","date":"2021-09-09 22:39:16.000000000","tz":-180},"subject":"Add the cinder-lvm charm to Openstack charms","message":"Add the cinder-lvm charm to Openstack charms\n\nThis charm implements an LVM backend for cinder, acting as\na subordinate charm.\n\nRequired-By: Ib2c481faa02a289c24c945378682e28b40fa6f33\nChange-Id: Ief7d7bc7fbb5b7c0028ead2d507d28700b54ac71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc7927f5b0047bf3fbf2331105b612788f00f6e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc7927f5b0047bf3fbf2331105b612788f00f6e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
