)]}'
{"id":"openstack%2Fproject-config~807838","triplet_id":"openstack%2Fproject-config~master~Ief7b70bf2d3fedcd5b193e7e718e3deb2a52fc89","project":"openstack/project-config","branch":"master","topic":"add-misc-charm-repos","hashtags":[],"change_id":"Ief7b70bf2d3fedcd5b193e7e718e3deb2a52fc89","subject":"Add openstack-loadbalancer charm and interfaces","status":"MERGED","created":"2021-09-08 08:48:05.000000000","updated":"2021-09-16 13:56:58.000000000","submitted":"2021-09-16 13:08:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807838-add-misc-charm-repos","meta_rev_id":"7cabb78bf4d78fb6759841cb281703ac7ea81c9e","_number":807838,"virtual_id_number":807838,"owner":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-16 13:08:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2021-09-08 10:49:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-09 22:31:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2021-09-16 12:58:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2021-09-16 12:58:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-08 08:55:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-08 10:49:07.000000000","updated_by":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"reviewer":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"state":"REVIEWER"},{"updated":"2021-09-09 22:31:34.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-09-16 12:58:22.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"e9951594c4c9da4cec81f147c47d32dd1f624c1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2021-09-08 08:48:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0a40ea2efa3c30306fe55f46accb6400898792b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 08:55:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/69461f8985114b34bb3519f04b7435a0 : SUCCESS in 6m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8534ace0964c4dd2b912875363ecfd42 : FAILURE in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a7329d50bbd217c9eb897dc0b246e1c0865e5e42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2021-09-08 09:33:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eaba4b0a5b6696c59b3cfb717974905e061ae94f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 09:46:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/afaf9dafd3194b91ab933fc4a092e4e0 : SUCCESS in 6m 08s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/230e839e5bc94026bb9235ca5a06bc15 : FAILURE in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"b9cb89315909a35bb40ea879a8f5bf3ca3aea8f4","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2021-09-08 09:59:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a48862ffdd4665bded2b3d8e632f359afc6e902d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 10:13:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/74b262ba8e8346a989d0c734f0d1e04a : SUCCESS in 6m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d6880afc08564fb486dc5f2f68f8e71e : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":2},{"id":"edad8b041ac2f5798bb9655f6e0a41cc65c1c56f","author":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"date":"2021-09-08 10:49:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"03428a51970dc604203c26f1b223264cea1d5cc1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-09 22:31:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6ef46802ffa8cb76d178e7be9fcdf5ad5225fcbc","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-09-16 12:58:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nTreating Ghanshyam\u0027s +1 as a +2 on this.","accounts_in_message":[],"_revision_number":2},{"id":"9ce349b14d4755e330ec2cd11d3dcd4a1ece1448","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 12:58:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fabd9b7aee22471d779f1cd86da3e96d9cedf14a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 13:08:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/65a6ff919c53431c94e25fc129f0edeb : SUCCESS in 6m 38s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2db29fa23bf64a019abe9f3481992245 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":2},{"id":"f2840e7b5060517406f0ebe6036fc0b433bb6b29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 13:08:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7cabb78bf4d78fb6759841cb281703ac7ea81c9e","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 13:56:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/259388172e394eda90a264af6aaebd46 : SUCCESS in 3m 19s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/454873ca99cb49c497d4d00289e12687 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"72bb30e04f73d5f7aba3ba1c49972d46d9bcf684","revisions":{"297334879349cdf1e248d8fb04d1b7c52e93d46b":{"kind":"REWORK","_number":1,"created":"2021-09-08 08:48:05.000000000","uploader":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"ref":"refs/changes/38/807838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/807838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/807838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/807838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/807838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/807838/1"}}},"commit":{"parents":[{"commit":"6207945a6b0f7c36bca7b69c8af17b66cd4f4b58","subject":"Remove the inap provider from nodepool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6207945a6b0f7c36bca7b69c8af17b66cd4f4b58"}]}],"author":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2021-09-08 08:46:47.000000000","tz":60},"committer":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2021-09-08 08:47:27.000000000","tz":60},"subject":"Add openstack-loadbalancer charm and interfaces","message":"Add openstack-loadbalancer charm and interfaces\n\nAdd new charm that provides loadbalancing to Openstack API services.\n\nAdd new interfaces which support the new loadbalancer and Ceph\ndashboard charms\n\nChange-Id: Ief7b70bf2d3fedcd5b193e7e718e3deb2a52fc89\nRequired-By: I804171fb4521372713c51cd99715643a7763866e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/297334879349cdf1e248d8fb04d1b7c52e93d46b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/297334879349cdf1e248d8fb04d1b7c52e93d46b"}]},"branch":"refs/heads/master"},"72bb30e04f73d5f7aba3ba1c49972d46d9bcf684":{"kind":"REWORK","_number":2,"created":"2021-09-08 09:33:15.000000000","uploader":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"ref":"refs/changes/38/807838/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/807838/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/807838/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/807838/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/807838/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/807838/2"}}},"commit":{"parents":[{"commit":"6207945a6b0f7c36bca7b69c8af17b66cd4f4b58","subject":"Remove the inap provider from nodepool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6207945a6b0f7c36bca7b69c8af17b66cd4f4b58"}]}],"author":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2021-09-08 08:46:47.000000000","tz":60},"committer":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2021-09-08 09:33:07.000000000","tz":60},"subject":"Add openstack-loadbalancer charm and interfaces","message":"Add openstack-loadbalancer charm and interfaces\n\nAdd new charm that provides loadbalancing to Openstack API services.\n\nAdd new interfaces which support the new loadbalancer and Ceph\ndashboard charms\n\nChange-Id: Ief7b70bf2d3fedcd5b193e7e718e3deb2a52fc89\nRequired-By: I804171fb4521372713c51cd99715643a7763866e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/72bb30e04f73d5f7aba3ba1c49972d46d9bcf684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/72bb30e04f73d5f7aba3ba1c49972d46d9bcf684"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
