)]}'
{"id":"openstack%2Fproject-config~811442","triplet_id":"openstack%2Fproject-config~master~I25d9bb7272edc3215840a53e5d79efe5d1fd7210","project":"openstack/project-config","branch":"master","topic":"cs9","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2021-10-28 06:58:44.000000000","reason":"Marios Andreou replied on the change"},"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2021-11-04 23:58:27.000000000","reason":"Change was submitted"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2021-11-04 23:58:27.000000000","reason":"Change was submitted"},"9976":{"account":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"last_update":"2021-10-26 11:06:52.000000000","reason":"Ronelle Landy replied on the change"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2021-10-26 10:24:58.000000000","reason":"Chandan Kumar replied on the change"},"10969":{"account":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"last_update":"2021-10-25 09:30:34.000000000","reason":"Shnaidman Sagi replied on the change"},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2021-11-04 23:43:25.000000000","reason":"Ian Wienand replied on the change"},"9311":{"account":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"last_update":"2021-11-04 23:58:27.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2021-11-04 14:48:04.000000000","reason":"removed on reply"}},"hashtags":["cs9"],"change_id":"I25d9bb7272edc3215840a53e5d79efe5d1fd7210","subject":"Add support for CentOS Stream 9 in nodepool elements","status":"MERGED","created":"2021-09-28 16:15:32.000000000","updated":"2021-11-05 00:05:23.000000000","submitted":"2021-11-04 23:58:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"811442-cs9","meta_rev_id":"537db8fa5fcb260603f8fb260933ba308f6a591a","_number":811442,"virtual_id_number":811442,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-04 23:58:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"recommended":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2021-11-05 00:05:23.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2021-11-05 00:05:23.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-04 23:43:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2021-11-04 14:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-04 23:43:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-28 16:25:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-29 14:07:46.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2021-09-30 08:10:49.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2021-10-19 06:33:22.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2021-10-19 06:33:22.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2021-10-19 06:33:22.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-10-19 06:33:22.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2021-10-25 08:27:18.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-10-26 11:07:11.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"state":"REVIEWER"},{"updated":"2021-10-28 23:13:23.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-10-29 08:41:35.000000000","updated_by":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"reviewer":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"state":"REVIEWER"},{"updated":"2021-11-04 08:44:54.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"dba0ef2396d54d321c710e4958371b3b2d1affe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-09-28 16:15:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68c6bbe3cb64d99289fab96d565c88ccac651bfe","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-09-28 16:15:50.000000000","message":"Hashtag added: cs9","accounts_in_message":[],"_revision_number":1},{"id":"a36cf79dd76a4b191a6ba407295cb5341086c9c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 16:25:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7a596a03355e472d8910544a60ac2bf0 : SUCCESS in 7m 47s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/fb7a38f4e8574f8caa0a0e960f83693e : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"a636055720b3c0c037fba7477664bb0c932acaef","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-10-25 05:29:25.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d41681e7978af69c8bb1e3ac0ce5685e4cceae4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-10-25 08:27:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"35d1277af742fa2b90f14ce9c2b3feb3d13b0265","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-10-25 09:30:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5098375c9d0015258e9faab3ab7fad92752015c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-10-25 13:58:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a7c9c90d5e8c17b32c3653543921e8509c9bc323","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 14:11:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b17d6b7ecc154799a101484c399c66d1 : SUCCESS in 7m 43s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/7d7a5fe00b034b34b079b979a8d0d62e : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":2},{"id":"ef0d5187fa1a73e77e32cacade865d269229417e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-10-26 07:47:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"58c2ec0a6fbd9416d29195b15f4d6fe25094b527","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-10-26 10:24:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ca0aa53cfabff9fc04c1a1e320204de656cee43d","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-10-26 11:06:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7663b47f2ee43d7b2584bfee3aedce8ec16f5c09","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-10-28 06:58:44.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4784ff34419355d6923e8f06d90a314b89325411","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-10-28 22:27:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"db2b060a744a4d25d55af82412f5a94cfe2e4d4f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-10-28 23:13:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"23f75e3e5adbaab286cebc0115814dc69298fd12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-10-29 08:06:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f3e73a8856ad61888466ac670f840cc4523d0ab5","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-10-29 08:06:15.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a34622d241af5e0a4a444e428b22b0828454655a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 08:13:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f42baa630ab34984b2caf5438b8739df : SUCCESS in 5m 44s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/5c5f666306e74a17a2137e47193ce20f : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":3},{"id":"e3f29c03495af9877ebf08425a25ffdd43b8d340","author":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"date":"2021-10-29 08:41:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"52264230e0ff494266dfdfc86de5ec9f9460bbad","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-10-29 08:57:49.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c3024c4f51f2febe102c6deb377df081b4731113","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-11-03 11:49:01.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"ef523281edb0f0e1a467db890647bae1a14cc950","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 12:03:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/343093277016408da32a5819bcf1ca3d : SUCCESS in 6m 33s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/589b77f787344441b2fc9b7d4d49499c : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":4},{"id":"76c41f955fbb006bbaef19b700d1aeaab9ebe2d1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-11-04 14:48:04.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7c35b796df1e273f24bcb520da932fa96bd1eea2","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-11-04 23:43:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"61cb0093d9f6a5ab71de6d85de0e13cb6b7dc7ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 23:43:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4192c28aed2e69871ab5b13cb59c40c7d4cfb589","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 23:58:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cce9fb149f494d03b761107163c982c0 : SUCCESS in 6m 11s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/9b20a6292e664dd6b3f7e09ad75463c1 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":4},{"id":"26628dc15f9f1f1550eadd7cf892194a55cfe98c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 23:58:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7fb1791858441c4b867c4585a06661e9a93b4af0","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 00:05:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/d445d99a1599456299fce1b14689bf56 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4e88e748bbfd877b45b4032ae899a6ef03fc05e6","revisions":{"c186987fca1b7ad8a60586ad9ab14f8d70102ef1":{"kind":"REWORK","_number":1,"created":"2021-09-28 16:15:32.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/42/811442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/42/811442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/811442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/811442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/811442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/42/811442/1"}}},"commit":{"parents":[{"commit":"c8a64e39dc50cdc7e598e3ecc576b8d12601032c","subject":"Merge \"Remove github3.py from our zuul config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c8a64e39dc50cdc7e598e3ecc576b8d12601032c"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-09-28 16:09:32.000000000","tz":0},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-09-28 16:15:27.000000000","tz":0},"subject":"Add support for CentOS Stream 9 in nodepool elements","message":"Add support for CentOS Stream 9 in nodepool elements\n\nThis patch is adding support for CentOS Stream 9 in elements\ninfra-package-needs and nodepool-base which are used in nodepool images.\n\n- Remove installation of ntpdate (it has been removed in CS9).\n- It uses RDO dependencies repo to install haveged until it\u0027s available in\n  EPEL9.\n- It adds package iptables-service which is not installed by default in CS9.\n\nNote that this patch has been tested together with dib patch in Depends-On.\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/811392\nChange-Id: I25d9bb7272edc3215840a53e5d79efe5d1fd7210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c186987fca1b7ad8a60586ad9ab14f8d70102ef1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c186987fca1b7ad8a60586ad9ab14f8d70102ef1"}]},"branch":"refs/heads/master"},"5f0278475a8404a3ac742be2d5bd66b9938cab87":{"kind":"REWORK","_number":2,"created":"2021-10-25 13:58:20.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/42/811442/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/42/811442/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/811442/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/811442/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/811442/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/42/811442/2"}}},"commit":{"parents":[{"commit":"c8a64e39dc50cdc7e598e3ecc576b8d12601032c","subject":"Merge \"Remove github3.py from our zuul config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c8a64e39dc50cdc7e598e3ecc576b8d12601032c"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-09-28 16:09:32.000000000","tz":0},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-10-25 13:58:18.000000000","tz":120},"subject":"Add support for CentOS Stream 9 in nodepool elements","message":"Add support for CentOS Stream 9 in nodepool elements\n\nThis patch is adding support for CentOS Stream 9 in elements\ninfra-package-needs and nodepool-base which are used in nodepool images.\n\n- Remove installation of ntpdate (it has been removed in CS9).\n- It uses RDO dependencies repo to install haveged until it\u0027s available in\n  EPEL9.\n- It adds package iptables-service which is not installed by default in CS9.\n\nNote that this patch has been tested together with dib patch in Depends-On.\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/811392\nChange-Id: I25d9bb7272edc3215840a53e5d79efe5d1fd7210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f0278475a8404a3ac742be2d5bd66b9938cab87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f0278475a8404a3ac742be2d5bd66b9938cab87"}]},"branch":"refs/heads/master"},"eb13473f40d37d260d944b4c03a8a6eb9e78a079":{"kind":"REWORK","_number":3,"created":"2021-10-29 08:06:06.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/42/811442/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/42/811442/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/811442/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/811442/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/811442/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/42/811442/3"}}},"commit":{"parents":[{"commit":"c8a64e39dc50cdc7e598e3ecc576b8d12601032c","subject":"Merge \"Remove github3.py from our zuul config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c8a64e39dc50cdc7e598e3ecc576b8d12601032c"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-09-28 16:09:32.000000000","tz":0},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-10-29 08:06:04.000000000","tz":120},"subject":"Add support for CentOS Stream 9 in nodepool elements","message":"Add support for CentOS Stream 9 in nodepool elements\n\nThis patch is adding support for CentOS Stream 9 in elements\ninfra-package-needs and nodepool-base which are used in nodepool images.\n\n- Remove installation of ntpdate (it has been removed in CS9).\n- It skips installation of haveged until it\u0027s available in EPEL9.\n- It maps package iptables to iptables-service.\n\nNote that this patch has been tested together with dib patch in Depends-On.\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/811392\nChange-Id: I25d9bb7272edc3215840a53e5d79efe5d1fd7210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb13473f40d37d260d944b4c03a8a6eb9e78a079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb13473f40d37d260d944b4c03a8a6eb9e78a079"}]},"branch":"refs/heads/master"},"4e88e748bbfd877b45b4032ae899a6ef03fc05e6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-11-03 11:49:01.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/42/811442/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/42/811442/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/811442/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/811442/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/811442/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/42/811442/4"}}},"commit":{"parents":[{"commit":"9f991cc09289ef39126e14394d985d1b799edf70","subject":"Merge \"kolla-cli: end gating for retirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9f991cc09289ef39126e14394d985d1b799edf70"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-09-28 16:09:32.000000000","tz":0},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-11-03 11:47:06.000000000","tz":60},"subject":"Add support for CentOS Stream 9 in nodepool elements","message":"Add support for CentOS Stream 9 in nodepool elements\n\nThis patch is adding support for CentOS Stream 9 in elements\ninfra-package-needs and nodepool-base which are used in nodepool images.\n\n- Remove installation of ntpdate (it has been removed in CS9).\n- It skips installation of haveged until it\u0027s available in EPEL9.\n- It maps package iptables to iptables-service.\n\nNote that this patch has been tested together with dib patch in Depends-On.\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/811392\nChange-Id: I25d9bb7272edc3215840a53e5d79efe5d1fd7210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4e88e748bbfd877b45b4032ae899a6ef03fc05e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4e88e748bbfd877b45b4032ae899a6ef03fc05e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
