)]}'
{"id":"openstack%2Fproject-config~813135","triplet_id":"openstack%2Fproject-config~master~Ifdb177d2ed75eab9e2acb984a9be785fba517d3e","project":"openstack/project-config","branch":"master","topic":"ubuntu-https","hashtags":[],"change_id":"Ifdb177d2ed75eab9e2acb984a9be785fba517d3e","subject":"nodepool: drop https for ubuntu","status":"MERGED","created":"2021-10-08 07:25:23.000000000","updated":"2021-10-09 12:27:18.000000000","submitted":"2021-10-09 12:21:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813135-ubuntu-https","meta_rev_id":"77056991041f541b9022ce5dc6e7a55580773d0d","_number":813135,"virtual_id_number":813135,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-09 12:21:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-10-08 19:11:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2021-10-08 07:51:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-09 12:14:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-09 12:14:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-08 07:36:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-08 07:51:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-10-08 19:11:07.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-10-09 12:14:33.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"d0922a3824ef38eb222423fce75aa860a81d178c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-10-08 07:25:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a8eb6709652968415b5be82f94c3236f16dc2cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-10-08 07:27:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c477ae5cbf2ee15684fd7e563206824c95b0fa63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 07:36:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/60c60fe43ae64e0dabae28fd1db8d4a7 : SUCCESS in 6m 11s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/618895c0be684165a90c38c59c39a576 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":2},{"id":"6958c787ae4d2c407f984fb932abf13a5f4891a2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-08 07:51:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aae2c4222d6cdfcb95525c10d5e4fdbaf82b5eac","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-10-08 19:11:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"436dac4b5c68f3182c0e87df4d2ae9ef370634c9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-10-09 12:14:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThis seems fine for now, and is a temporary measure until we get the docker base image bootstrapping solution fully in place, as that should address it for us anyway.","accounts_in_message":[],"_revision_number":2},{"id":"03c1f56c9a92982dbbf2e8ffdb8e4b0ef663e4cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 12:14:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"98661dc23b5016a3544c9288844107cddd9daae4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 12:21:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c95c889533c54e63a94e944cb0e71d31 : SUCCESS in 5m 25s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/21ef9026d6634a048e86d9f9b4b64ac8 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ec2478b0b2cd58ff8dd84ae2743eba04a479aab1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 12:21:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"77056991041f541b9022ce5dc6e7a55580773d0d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 12:27:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/b08a4a0000964f7088f0cff100697d14 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"05adea62b2324ac978ceb2c4bcd9c07e83aa60c2","revisions":{"7d3070a972dc352bedddd10a812b48da33c4b0c7":{"kind":"REWORK","_number":1,"created":"2021-10-08 07:25:23.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/35/813135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/813135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/813135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/813135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/813135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/813135/1"}}},"commit":{"parents":[{"commit":"e2c1317e11351d87f0d2fde61fbdc2dab1cb741c","subject":"Update gerrit image dep for jeepyb triggered builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2c1317e11351d87f0d2fde61fbdc2dab1cb741c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-08 07:22:55.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-08 07:22:55.000000000","tz":660},"subject":"nodepool: drop https for ubuntu","message":"nodepool: drop https for ubuntu\n\nLE recently expired it\u0027s root certificate.  Unfortunately, the base\nUbuntu distros (up to focal) use a 2019 ca-certificates file that\nhasn\u0027t accounted for this.  Even more unfortunately, deboostrap can\u0027t\nuse the updates repo to get a more recent certificate package.\n\nThe end result is that the build chroot can\u0027t contact the mirrors via\nhttp to actually pull the updated certificates package from them.\n\nTo work around this, use http:// for the mirror location, so the\ninitial upgrade doesn\u0027t try to use https.\n\nChange-Id: Ifdb177d2ed75eab9e2acb984a9be785fba517d3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d3070a972dc352bedddd10a812b48da33c4b0c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d3070a972dc352bedddd10a812b48da33c4b0c7"}]},"branch":"refs/heads/master"},"05adea62b2324ac978ceb2c4bcd9c07e83aa60c2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-08 07:27:49.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/35/813135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/813135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/813135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/813135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/813135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/813135/2"}}},"commit":{"parents":[{"commit":"e2c1317e11351d87f0d2fde61fbdc2dab1cb741c","subject":"Update gerrit image dep for jeepyb triggered builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2c1317e11351d87f0d2fde61fbdc2dab1cb741c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-08 07:22:55.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-08 07:26:58.000000000","tz":660},"subject":"nodepool: drop https for ubuntu","message":"nodepool: drop https for ubuntu\n\nLE recently expired it\u0027s root certificate.  Unfortunately, the base\nUbuntu distros (up to focal) use a 2019 ca-certificates package that\nhasn\u0027t accounted for this.  Even more unfortunately, deboostrap can\u0027t\nuse the updates repo to get a more recent certificate package when\nbuilding the initial chroot.\n\nThe end result is that the chroot can\u0027t contact the mirrors via https\nto actually pull the updated certificates package from them, breaking\nthe build.\n\nTo work around this use http:// for the mirror location.\n\nChange-Id: Ifdb177d2ed75eab9e2acb984a9be785fba517d3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/05adea62b2324ac978ceb2c4bcd9c07e83aa60c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/05adea62b2324ac978ceb2c4bcd9c07e83aa60c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
