)]}'
{"id":"openstack%2Fproject-config~813876","triplet_id":"openstack%2Fproject-config~master~I73cab4c0c581e70fdb3b972c0fdc1f8eed637583","project":"openstack/project-config","branch":"master","topic":"pause-f34","hashtags":[],"change_id":"I73cab4c0c581e70fdb3b972c0fdc1f8eed637583","subject":"Pause fedora-34 to debug network problems","status":"ABANDONED","created":"2021-10-13 17:12:39.000000000","updated":"2021-10-13 17:57:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"861004726f5fd740925ee2f091a801799e0d6993","_number":813876,"virtual_id_number":813876,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-13 17:26:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-13 17:26:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0396e0baf8df4de3ec35b3da1968fe51c5ad4826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-10-13 17:12:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11bead3213d8ded83e639686b3a8416201fd28c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 17:26:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/182d5a5e435b4605b440c1f1a66e44ab : SUCCESS in 6m 17s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/10c3d377b46b4c5383a72aca124d2322 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"861004726f5fd740925ee2f091a801799e0d6993","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-10-13 17:57:09.000000000","message":"Abandoned\n\nTurns out the old image doesn\u0027t work either so this won\u0027t help","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a6212c10f08553918b4e21982f4d5c490dede6b","revisions":{"1a6212c10f08553918b4e21982f4d5c490dede6b":{"kind":"REWORK","_number":1,"created":"2021-10-13 17:12:39.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/76/813876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/813876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/813876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/813876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/813876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/813876/1"}}},"commit":{"parents":[{"commit":"9998b4da6b40568483d8786f1b4a478147e9d105","subject":"Add creating branch ACL to ansible-collections-openstack repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9998b4da6b40568483d8786f1b4a478147e9d105"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-10-13 17:11:58.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-10-13 17:11:58.000000000","tz":-420},"subject":"Pause fedora-34 to debug network problems","message":"Pause fedora-34 to debug network problems\n\nWe seem to have network setup issues with fedora-34s latest images.\nPause the image so that we can delete the most recent build then figure\nit out from there without nodepool replacing things with broken images\nagain.\n\nChange-Id: I73cab4c0c581e70fdb3b972c0fdc1f8eed637583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1a6212c10f08553918b4e21982f4d5c490dede6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1a6212c10f08553918b4e21982f4d5c490dede6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
