)]}'
{"id":"openstack%2Fproject-config~814381","triplet_id":"openstack%2Fproject-config~master~I6afb3c1ea67fce8c9f9d58295e3a393d1b35c6ea","project":"openstack/project-config","branch":"master","topic":"add-ttx-ptg-op","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-10-18 12:08:33.000000000","reason":"Change was submitted"},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2021-10-18 12:08:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6afb3c1ea67fce8c9f9d58295e3a393d1b35c6ea","subject":"Add ttx as OP to #openinfra-events","status":"MERGED","created":"2021-10-18 09:07:08.000000000","updated":"2021-10-18 12:39:25.000000000","submitted":"2021-10-18 12:08:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814381-add-ttx-ptg-op","meta_rev_id":"fddbbf3c9429a2818c574f7dd30832c9f8aeb351","_number":814381,"virtual_id_number":814381,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-18 12:08:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2021-10-18 12:39:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-18 11:44:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-18 11:44:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-18 09:13:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-10-18 09:25:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-18 11:44:30.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"b618fe74f7ada4e24296c734c66796355767a559","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-10-18 09:07:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3695907f5800677129243aa743fa4d3511e1a778","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-18 09:13:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3bee993f677c12ec81e8454f44aa6b2072d7eb15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 09:25:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/498f6744d3b14dda99c9754bfa385893 : FAILURE in 7m 34s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/06a4e8ad005048669c57f50055babc2a : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d3c5d3fcf24bc19203793a3f93b74bf4fff1c3ae","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-10-18 09:40:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd10c2710328812c651dfea19b4c6cc24cde5ee6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 09:56:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9915bb81191c485d81da3bda3ee5d9d8 : FAILURE in 6m 47s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/00cad39c4c214a3cbef34761257ce7e9 : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1d4fca12fea95ad2a717d7a5591083d8941e8a58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-18 11:36:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"231370af880c45281443a72fc5bfea2f4906927b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-10-18 11:44:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b57c53b1a9ac4b290a9bccec7c139e800afffd03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 11:48:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/501ec10b01784d9899cbd53ea340b2e9 : SUCCESS in 6m 32s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/ce50a22c474d46f895f395f3c35e62da : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":2},{"id":"92c9300f227bdfa2dd13e724f35becdefdb42011","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 11:49:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ead4d468d389834dae6f92f36ab1de21bef40841","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 12:08:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/117b18b7218a463aaf5bf7229d92628c : SUCCESS in 5m 59s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/1e643b15a17849b49ba7490998340b4b : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":2},{"id":"d27b51a146f8391c89b7191dc2c895893a4601a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 12:08:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"652e0b7dfdcec10d1265fef473f3282e388e4c17","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 12:39:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-run-accessbot https://zuul.opendev.org/t/openstack/build/c3b312b99ebe4bbcaa11a32c34441411 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a7673aa534de3138c2e857d520b3ab4faae6e04c","revisions":{"b0a212467975b441c4c9375001cb8c199a114f2c":{"kind":"REWORK","_number":1,"created":"2021-10-18 09:07:08.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/81/814381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/81/814381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/814381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/814381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/814381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/81/814381/1"}}},"commit":{"parents":[{"commit":"611a5e0f665ba0d33048055fd4e780f0d4d018b8","subject":"Merge \"Move ansible-role-refstack-client from x/ to openinfra/\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/611a5e0f665ba0d33048055fd4e780f0d4d018b8"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2021-10-18 09:05:44.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2021-10-18 09:06:56.000000000","tz":120},"subject":"Add ttx as OP to #openinfra-events","message":"Add ttx as OP to #openinfra-events\n\nIn order to help drive the ptgbot during the PTG event, give\nttx OP rights on #openinfra-events channel.\n\nChange-Id: I6afb3c1ea67fce8c9f9d58295e3a393d1b35c6ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b0a212467975b441c4c9375001cb8c199a114f2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b0a212467975b441c4c9375001cb8c199a114f2c"}]},"branch":"refs/heads/master"},"a7673aa534de3138c2e857d520b3ab4faae6e04c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-18 11:36:21.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/81/814381/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/81/814381/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/814381/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/814381/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/814381/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/81/814381/2"}}},"commit":{"parents":[{"commit":"6ca813f6c836de74fd88bd2edb89ce85d903b284","subject":"Fix project-config testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6ca813f6c836de74fd88bd2edb89ce85d903b284"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2021-10-18 09:05:44.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2021-10-18 11:36:21.000000000","tz":0},"subject":"Add ttx as OP to #openinfra-events","message":"Add ttx as OP to #openinfra-events\n\nIn order to help drive the ptgbot during the PTG event, give\nttx OP rights on #openinfra-events channel.\n\nChange-Id: I6afb3c1ea67fce8c9f9d58295e3a393d1b35c6ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a7673aa534de3138c2e857d520b3ab4faae6e04c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a7673aa534de3138c2e857d520b3ab4faae6e04c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
