)]}'
{"id":"openstack%2Fproject-config~814401","triplet_id":"openstack%2Fproject-config~master~I439d1f167b5893a134a5b628ebc00fd6617e0cbd","project":"openstack/project-config","branch":"master","topic":"fix-yaml-load","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-10-18 11:35:58.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"}},"hashtags":[],"change_id":"I439d1f167b5893a134a5b628ebc00fd6617e0cbd","subject":"Fix project-config testing","status":"MERGED","created":"2021-10-18 10:26:31.000000000","updated":"2021-10-18 11:51:52.000000000","submitted":"2021-10-18 11:48:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814401-fix-yaml-load","meta_rev_id":"e7e1879a7ac83dee03bdea9a962b6cf08a2ba231","_number":814401,"virtual_id_number":814401,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-18 11:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2021-10-18 11:35:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2021-10-18 11:35:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-18 10:39:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc0f1cad30f462970dc93f0be83986f1d690c8af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-18 10:26:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbe0cd7b178542b7a7ef08b24f8f0c11ea315c49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 10:39:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/2f68576ed5ab422097f9dd4617abf4eb : SUCCESS in 4m 25s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3a95abba685a45f996cec4fe2458477b : SUCCESS in 5m 41s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/9fb9d1bb2b134f66a31f51a7a3453a3a : SUCCESS in 4m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/381a9847f13041798819e4d911f2aa24 : FAILURE in 4m 33s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/d010e7f2bdca45f9bab0c4877a50dadc : SUCCESS in 4m 34s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/4ac718d8d6ea474899f8e9a7c411d522 : SUCCESS in 4m 49s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/a890c7d60f1e4bd2966add97506f748c : SUCCESS in 6m 56s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/1eb29f46c9c642b9aa96294ac886e1ee : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e6ea6a9aae8a04ec30db86993d29b152570a9542","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-18 11:10:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0b0d6e8e4d05da7bd3e664b493d12c2e2073f0f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-18 11:23:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8b35297e8388eb02c8d613fd406e4fbd3bba2d69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 11:34:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/73a5566dc3324ae3b7e52bfd869beaa0 : SUCCESS in 4m 16s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/17d1890563554d67b68c6ce9d555cf38 : SUCCESS in 6m 20s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/8ef4708432634579ad51e9909df1cf24 : SUCCESS in 4m 12s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7af281e364fc4993a919c76879a8afc6 : SUCCESS in 4m 33s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/a170f0f172254492b518a61906cf2a54 : SUCCESS in 4m 23s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/6502c427d0ed42db8e1280e75522fe85 : SUCCESS in 4m 15s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/b4854cdc12c54c808f7f54b6067c0c35 : SUCCESS in 6m 36s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/0e2a4cc352654efe91ef9b66f8d1b65f : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":3},{"id":"9e8358954e3c3711dd060489c2824e287d7a03fe","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-18 11:35:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"183058853259060f4da52bc12ca86271f36397d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 11:36:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e0f3f23ad21f31b74dc4009cb9fef8b3d1f8157e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 11:48:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/78e616ede2964e01a7b9583fc74cd9f5 : SUCCESS in 3m 59s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e1874a62518b47d097f7146107a94952 : SUCCESS in 6m 12s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/272bf01b89a0449bac7dcad20fffbf6e : SUCCESS in 4m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ef6ae8e614ef4d25850c3f5d21c6a2b2 : SUCCESS in 5m 42s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/d18d3a66676c40e396bca70ed578c235 : SUCCESS in 4m 27s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/b5253461faf74cf8b038bbde286df51a : SUCCESS in 4m 21s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/0bbfb4b481be4761bc0a24ef24da94cc : SUCCESS in 6m 27s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/21c042b23635416487b94e7e6b483505 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":3},{"id":"08939f1eedd541c07997cb2b532e9d899427fe02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 11:48:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e7e1879a7ac83dee03bdea9a962b6cf08a2ba231","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 11:51:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- promote-infra-index https://zuul.opendev.org/t/openstack/build/f2a30a92fd13441e8bb31bd9f32f2766 : SUCCESS in 1m 11s\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/12a57793380d463598744c1027af6a49 : SUCCESS in 1m 20s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/4b0e401321654dacb0bfdfb58b958111 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6ca813f6c836de74fd88bd2edb89ce85d903b284","revisions":{"9d54ee12b3c74afb1d6f2a56bdcef9aea1e59bd2":{"kind":"REWORK","_number":1,"created":"2021-10-18 10:26:31.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/01/814401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/814401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/814401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/814401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/814401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/814401/1"}}},"commit":{"parents":[{"commit":"611a5e0f665ba0d33048055fd4e780f0d4d018b8","subject":"Merge \"Move ansible-role-refstack-client from x/ to openinfra/\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/611a5e0f665ba0d33048055fd4e780f0d4d018b8"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-10-18 10:25:31.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-10-18 10:26:29.000000000","tz":120},"subject":"Fix use of yaml.load()","message":"Fix use of yaml.load()\n\nThe use of this function has been deprecated for a long time[0]. With\nPyYAML\u003d\u003d6.0 the call is now failing, so replace it with the safe\nversion.\n\n[0] https://msg.pyyaml.org/load\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I439d1f167b5893a134a5b628ebc00fd6617e0cbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9d54ee12b3c74afb1d6f2a56bdcef9aea1e59bd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9d54ee12b3c74afb1d6f2a56bdcef9aea1e59bd2"}]},"branch":"refs/heads/master"},"d398199458f95a181564c2437d3fcfcc99dcf32a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-18 11:10:16.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/01/814401/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/814401/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/814401/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/814401/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/814401/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/814401/2"}}},"commit":{"parents":[{"commit":"30f585a3a6ba0c79ae1f7d7b9fe05c0845839a72","subject":"Fix the renaming of ansible-role-refstack-client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/30f585a3a6ba0c79ae1f7d7b9fe05c0845839a72"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-10-18 10:25:31.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-10-18 11:08:03.000000000","tz":120},"subject":"Fix use of yaml.load()","message":"Fix use of yaml.load()\n\nThe use of this function has been deprecated for a long time[0]. With\nPyYAML\u003d\u003d6.0 the call is now failing, so replace it with the safe\nversion.\n\n[0] https://msg.pyyaml.org/load\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I439d1f167b5893a134a5b628ebc00fd6617e0cbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d398199458f95a181564c2437d3fcfcc99dcf32a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d398199458f95a181564c2437d3fcfcc99dcf32a"}]},"branch":"refs/heads/master"},"6ca813f6c836de74fd88bd2edb89ce85d903b284":{"kind":"REWORK","_number":3,"created":"2021-10-18 11:23:48.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/01/814401/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/814401/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/814401/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/814401/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/814401/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/814401/3"}}},"commit":{"parents":[{"commit":"611a5e0f665ba0d33048055fd4e780f0d4d018b8","subject":"Merge \"Move ansible-role-refstack-client from x/ to openinfra/\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/611a5e0f665ba0d33048055fd4e780f0d4d018b8"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-10-18 11:05:07.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-10-18 11:23:29.000000000","tz":120},"subject":"Fix project-config testing","message":"Fix project-config testing\n\nThis is a combination of 2 commits:\n\nFix the renaming of ansible-role-refstack-client\n\nIn [0] that project was moved from x/ to openinfra/, but the patch\nmissed that at the same time refstack was also moved, changing the\npath for the ACL.[1]\n\n[0] https://review.opendev.org/c/openstack/project-config/+/765787\n[1] https://review.opendev.org/c/openstack/project-config/+/808479\n\nFix use of yaml.load()\n\nThe use of this function has been deprecated for a long time[2]. With\nPyYAML\u003d\u003d6.0 the call is now failing, so replace it with the safe\nversion.\n\n[2] https://msg.pyyaml.org/load\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I439d1f167b5893a134a5b628ebc00fd6617e0cbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6ca813f6c836de74fd88bd2edb89ce85d903b284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6ca813f6c836de74fd88bd2edb89ce85d903b284"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
