)]}'
{"id":"openstack%2Fproject-config~814409","triplet_id":"openstack%2Fproject-config~master~I2bac8f901ecd3f74b70aec1f8d29c45d10506900","project":"openstack/project-config","branch":"master","topic":"fix-yaml-load","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-10-18 11:24:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2bac8f901ecd3f74b70aec1f8d29c45d10506900","subject":"Fix the renaming of ansible-role-refstack-client","status":"ABANDONED","created":"2021-10-18 11:10:16.000000000","updated":"2021-10-18 11:24:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e587b1a96d21138bc5598074ac815b4f9c013b3d","_number":814409,"virtual_id_number":814409,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-18 11:18:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-18 11:18:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7f5ba3f75bc0b6ffc4030a58673fe38a2e112493","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-18 11:10:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cccde3dbd897a3cbcad4723cc15cc236feee6137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 11:18:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bc365d08f3e34ecc8bd9d6f82674c8f6 : FAILURE in 6m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a07627e1278d4f228c7fe6b3055875cb : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e587b1a96d21138bc5598074ac815b4f9c013b3d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-18 11:24:55.000000000","message":"Abandoned\n\nMerged into https://review.opendev.org/c/openstack/project-config/+/814401 to fix two issues at the same time","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30f585a3a6ba0c79ae1f7d7b9fe05c0845839a72","revisions":{"30f585a3a6ba0c79ae1f7d7b9fe05c0845839a72":{"kind":"REWORK","_number":1,"created":"2021-10-18 11:10:16.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/09/814409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/09/814409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/814409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/814409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/814409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/09/814409/1"}}},"commit":{"parents":[{"commit":"611a5e0f665ba0d33048055fd4e780f0d4d018b8","subject":"Merge \"Move ansible-role-refstack-client from x/ to openinfra/\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/611a5e0f665ba0d33048055fd4e780f0d4d018b8"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-10-18 11:05:07.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-10-18 11:07:37.000000000","tz":120},"subject":"Fix the renaming of ansible-role-refstack-client","message":"Fix the renaming of ansible-role-refstack-client\n\nIn [0] that project was moved from x/ to openinfra/, but the patch\nmissed that at the same time refstack was also moved, changing the\npath for the ACL.[1]\n\n[0] https://review.opendev.org/c/openstack/project-config/+/765787\n[1] https://review.opendev.org/c/openstack/project-config/+/808479\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I2bac8f901ecd3f74b70aec1f8d29c45d10506900\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/30f585a3a6ba0c79ae1f7d7b9fe05c0845839a72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/30f585a3a6ba0c79ae1f7d7b9fe05c0845839a72"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
