)]}'
{"id":"openstack%2Fproject-config~815497","triplet_id":"openstack%2Fproject-config~master~Ic4a8c47b171546496503837c9e918d69e2bb176c","project":"openstack/project-config","branch":"master","topic":"dpl-validate-ptl","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2021-10-28 10:23:31.000000000","reason":"Thierry Carrez replied on the change"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-06-17 18:53:50.000000000","reason":"Change was abandoned"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2026-06-17 18:53:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic4a8c47b171546496503837c9e918d69e2bb176c","subject":"Allow to check release approval for DPL teams","status":"ABANDONED","created":"2021-10-26 13:08:08.000000000","updated":"2026-06-17 18:53:50.000000000","total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"b48e36137325555a587d9150019edb9c6964e06c","_number":815497,"virtual_id_number":815497,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-08 15:22:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":-1,"date":"2023-09-28 12:39:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2023-09-28 15:06:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-26 13:08:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-10-26 13:08:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-10-26 13:26:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-27 07:11:56.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2022-07-08 15:23:56.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"0892fdd0243fcd5804fa743d1a4b9a2fcbd05962","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-26 13:08:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72f0e2761f372bf99483dd1b9dbf91e2729e5a03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-26 13:09:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd806d444860a9129910e5ccabd428f1bb5e2ad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 13:26:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a99793f833ca4639b01687e19a012a04 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":2},{"id":"10ca23a9b9953975df86c7dc4b98764a588471a8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-10-27 07:11:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4abc39350dd59d6f3cb2177c09234f3227cf4785","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-10-28 10:23:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"45c699c42f7da6f818437a1d9e8f1c73002264ea","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-11-02 14:43:23.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"23ded5181458079fd414e2df7f1988f59ebbea58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-08 15:06:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1daae1b2402bea9eadf639c0623c70e28a20cccd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-08 15:10:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f3812289b554772cdcd9bc6ee47ce4fdbe9563b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 15:22:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7b97d9e20c840f6adfcb25d2b850b76\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/3079c8587eae4f2e96f169e1784b4c74 : SUCCESS in 3m 00s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c451ec588cde4da49dfc0b6a26f31b5c : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":3},{"id":"e997f7943c3b26599bcca0ec1b8e968986e4c89c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-07-08 15:23:56.000000000","message":"Patch Set 3: Code-Review+2\n\nIs it possible for a team to have both a release liaison and a PTL recorded in governance? I guess the current assumption is that if a PTL exists then any additional release liaison will be listed in the releases repo, but if there is no PTL then there must be a release liaison in governance.\n\nMaybe it\u0027s worth having a separate discussion about moving all liaison tracking into governance, whether or not the team uses a DPL model.","accounts_in_message":[],"_revision_number":3},{"id":"d5b0518e4c470cf2f4a17c3ffc15dd1dfef5f789","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-28 12:36:38.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a6e199ae29ec6397e274f3df8161456353420ad2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-28 12:39:48.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"86a887f8ae56f8f9f027d828cfb91c04a9039c0b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-28 15:06:33.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"b48e36137325555a587d9150019edb9c6964e06c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-17 18:53:50.000000000","message":"Abandoned\n\nthis is doubly obsolete now since release liaisons are now tracked in the governance repo, too","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f2a5c9a1c929f4ba912066a6581b7d472eb93230","revisions":{"e49ba0b58cc53965295aa7ed4c31951723b86eed":{"kind":"REWORK","_number":1,"created":"2021-10-26 13:08:08.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/97/815497/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/97/815497/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/815497/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/815497/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/815497/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/97/815497/1"}}},"commit":{"parents":[{"commit":"b472c9deb818d96a2e40dff11fa1a4f946bdf13b","subject":"Merge \"Mirror newly added charms to GitHub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b472c9deb818d96a2e40dff11fa1a4f946bdf13b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-10-26 13:03:36.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-10-26 13:07:37.000000000","tz":120},"subject":"Allow to check release approval for DPL teams","message":"Allow to check release approval for DPL teams\n\nThese teams are PTL-less and not necessary define their\nrelease liaison openstack/release/data/release_liaison.yaml.\n\nThe problem is that their release liaison approval is not\nnecessarly detected.\n\nThese changes allow to consider these release liaisons folks.\n\nChange-Id: Ic4a8c47b171546496503837c9e918d69e2bb176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e49ba0b58cc53965295aa7ed4c31951723b86eed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e49ba0b58cc53965295aa7ed4c31951723b86eed"}]},"branch":"refs/heads/master"},"f2de286150fcf0434cf5381e048b1d52dfe5c953":{"kind":"REWORK","_number":2,"created":"2021-10-26 13:09:29.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/97/815497/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/97/815497/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/815497/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/815497/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/815497/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/97/815497/2"}}},"commit":{"parents":[{"commit":"b472c9deb818d96a2e40dff11fa1a4f946bdf13b","subject":"Merge \"Mirror newly added charms to GitHub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b472c9deb818d96a2e40dff11fa1a4f946bdf13b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-10-26 13:03:36.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-10-26 13:09:16.000000000","tz":120},"subject":"Allow to check release approval for DPL teams","message":"Allow to check release approval for DPL teams\n\nThese teams are PTL-less and not necessary define their\nrelease liaison openstack/release/data/release_liaison.yaml.\n\nThe problem is that their release liaison approval is not\nnecessarly detected.\n\nThese changes allow to consider these release liaisons folks.\n\nChange-Id: Ic4a8c47b171546496503837c9e918d69e2bb176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2de286150fcf0434cf5381e048b1d52dfe5c953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2de286150fcf0434cf5381e048b1d52dfe5c953"}]},"branch":"refs/heads/master"},"f2a5c9a1c929f4ba912066a6581b7d472eb93230":{"kind":"REWORK","_number":3,"created":"2022-07-08 15:06:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/97/815497/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/97/815497/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/815497/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/815497/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/815497/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/97/815497/3"}}},"commit":{"parents":[{"commit":"b472c9deb818d96a2e40dff11fa1a4f946bdf13b","subject":"Merge \"Mirror newly added charms to GitHub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b472c9deb818d96a2e40dff11fa1a4f946bdf13b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-10-26 13:03:36.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-08 15:04:08.000000000","tz":120},"subject":"Allow to check release approval for DPL teams","message":"Allow to check release approval for DPL teams\n\nThese teams are PTL-less and not necessary define their\nrelease liaison openstack/release/data/release_liaison.yaml.\n\nThe problem is that their release liaison approval is not\nnecessarly detected.\n\nThese changes allow to consider these release liaisons folks.\n\nChange-Id: Ic4a8c47b171546496503837c9e918d69e2bb176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2a5c9a1c929f4ba912066a6581b7d472eb93230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2a5c9a1c929f4ba912066a6581b7d472eb93230"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
