)]}'
{"id":"openstack%2Fproject-config~817507","triplet_id":"openstack%2Fproject-config~master~I3394c227fbbc07f1e797da7274781d0bc086e7ed","project":"openstack/project-config","branch":"master","topic":"retire-training-labs","attention_set":{},"removed_from_attention_set":{"6547":{"account":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"last_update":"2021-11-20 17:06:12.000000000","reason":"Andreas Jaeger replied on the change"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2021-11-20 17:16:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3394c227fbbc07f1e797da7274781d0bc086e7ed","subject":"Retire training-labs: remove project infra","status":"MERGED","created":"2021-11-11 01:37:27.000000000","updated":"2021-11-20 17:53:17.000000000","submitted":"2021-11-20 17:16:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"817507-retire-training-labs","meta_rev_id":"95f4d6eb2f3af264c03ffd06120691d077e0938a","_number":817507,"virtual_id_number":817507,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-20 17:16:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-11-19 20:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2021-11-20 17:06:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2021-11-20 17:06:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-11 01:48:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-15 11:23:57.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-11-17 17:52:32.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"7d3592ba71726e0b6379e3f4e1d90a5948e6d6c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-11 01:37:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a81c0c2ea90313900d5eceb8f9ac3c30d7a8fc0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 01:48:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1f3a1e9a91b3412da36fe9b0603a9f93 : SUCCESS in 5m 55s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fd957fe13edc467290ac28c20c7721d3 : SUCCESS in 4m 47s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/ab53750e1f2e4fc9a970ac8f17adc27d : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ff2517d9d15e114079ba7d54958e38052641e04e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-11-15 11:23:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"018c7d1603bd59274e9d751641b6b623c44e0e6a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-17 17:48:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4955b133ae138cd3783968bb93d7b16c894ed60a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-17 17:50:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5e8eebb32433d04354f8480b1f00db0168340fd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 17:51:56.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/training-labs on branch stable/ocata.  The error was:\n\n  Job publish-training-labs-scripts not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs\n      post:\n        jobs:\n          - publish-training-labs-scripts\n\n  in \"openstack/training-labs/.zuul.yaml@stable/ocata\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"2fef3781ff1e91ee11e75c61792db012938b1a27","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-11-17 17:52:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0acc24fe925364ce0b603c3b3f2f1d958c5259cb","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-11-17 17:53:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e7f8a9fafcf4f9f43b3ee4fae189e6fdabde96e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-17 18:00:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b247511737b9b59eefdae7ab1849b5f1ca2f906","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-11-18 07:17:51.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff06d1229c2425ab7fde11624730de0e40419fa6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-18 17:38:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aa3e8854f139166616cd1596e8e9b4b8473d920f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-19 20:03:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"791d1a5486f1db7608f059e75ad1b14cddeb0033","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 20:04:18.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/training-labs on branch stable/ocata.  The error was:\n\n  Job publish-training-labs-scripts not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs\n      post:\n        jobs:\n          - publish-training-labs-scripts\n\n  in \"openstack/training-labs/.zuul.yaml@stable/ocata\", line 1, column 3","accounts_in_message":[],"_revision_number":3},{"id":"dc3eb93bccb6a2735b8c261d770f10a0dd2e7218","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-19 20:07:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"73b6a440d0f202b503721d0caa0b1c876733ec89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 20:20:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2e026f1b11044d878aff29002a21de94 : SUCCESS in 6m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2052fba5598e45c6ac1ba1a27344b74c : SUCCESS in 5m 16s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/1368578734c14d519c791316e95d1fb1 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":4},{"id":"7b0dbf258e62476536a6ec47c0d99af3a86836f1","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-11-19 20:41:53.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f263fea15a0d4b8120715aa6fc090dd2ca208c3c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-11-20 17:06:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8c53e39092fe5447e56321553b117cabe5a23305","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 17:06:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c20ab5028b02243f4162b1e9e3c6433c1451f4d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 17:16:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4e4fbbab8eb84584a2127d644dd79ab4 : SUCCESS in 8m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/02c45aad3d8f4d83825b1566762510ee : SUCCESS in 4m 54s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/cdb4322fa9f046d0b7ba026663f19ab6 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":4},{"id":"fb44a7f1970a0e385b2079343b8a1f0b4b19648d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 17:16:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"95f4d6eb2f3af264c03ffd06120691d077e0938a","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 17:53:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/00d5145222804c1d8ef3524097b44098 : SUCCESS in 4m 53s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/af7dad6545ad4611bdaaf2258050cf1e : SUCCESS in 8m 06s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/0cba2c757e3a4ed3a1a1de444c47b4cb : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4a77159060a125eb54ac0ef3600ed35cb97f599e","revisions":{"db5a45bf706f9d8ac1b2ad4e83e5226114e03de2":{"kind":"REWORK","_number":1,"created":"2021-11-11 01:37:27.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/07/817507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/07/817507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/817507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/817507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/817507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/07/817507/1"}}},"commit":{"parents":[{"commit":"55ffda006d8c774ffdc874e08b35209d9f9a819b","subject":"Retire training-labs: add noop-jobs, update acl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/55ffda006d8c774ffdc874e08b35209d9f9a819b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-11 01:36:06.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-11 01:36:06.000000000","tz":-360},"subject":"Retire training-labs: remove project infra","message":"Retire training-labs: remove project infra\n\nDepends-On: https://review.opendev.org/c/openstack/training-labs/+/817505\nChange-Id: I3394c227fbbc07f1e797da7274781d0bc086e7ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/db5a45bf706f9d8ac1b2ad4e83e5226114e03de2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/db5a45bf706f9d8ac1b2ad4e83e5226114e03de2"}]},"branch":"refs/heads/master"},"4ff4221788eed41b7a342e698571da2497ebeae2":{"kind":"REWORK","_number":2,"created":"2021-11-17 17:50:05.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/07/817507/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/07/817507/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/817507/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/817507/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/817507/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/07/817507/2"}}},"commit":{"parents":[{"commit":"8f036503536ddbf66d9bb7863db6c6e25366674d","subject":"Merge \"Iotronic-pythonclient and Iotronic-UI update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8f036503536ddbf66d9bb7863db6c6e25366674d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-11 01:36:06.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-17 17:49:51.000000000","tz":-360},"subject":"Retire training-labs: remove project infra","message":"Retire training-labs: remove project infra\n\nDepends-On: https://review.opendev.org/c/openstack/training-labs/+/817505\nChange-Id: I3394c227fbbc07f1e797da7274781d0bc086e7ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ff4221788eed41b7a342e698571da2497ebeae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ff4221788eed41b7a342e698571da2497ebeae2"}]},"branch":"refs/heads/master"},"88185508320f9196cbf1d32655a7ca0efea43812":{"kind":"REWORK","_number":3,"created":"2021-11-19 20:03:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/07/817507/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/07/817507/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/817507/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/817507/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/817507/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/07/817507/3"}}},"commit":{"parents":[{"commit":"edf88041b3ff51209289fd0d1bb7c77e28c5da68","subject":"Merge \"Retire puppet-senlin - Step 3: Remove Project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/edf88041b3ff51209289fd0d1bb7c77e28c5da68"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-11 01:36:06.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-19 20:03:27.000000000","tz":-360},"subject":"Retire training-labs: remove project infra","message":"Retire training-labs: remove project infra\n\nDepends-On: https://review.opendev.org/c/openstack/training-labs/+/817505\nChange-Id: I3394c227fbbc07f1e797da7274781d0bc086e7ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88185508320f9196cbf1d32655a7ca0efea43812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88185508320f9196cbf1d32655a7ca0efea43812"}]},"branch":"refs/heads/master"},"4a77159060a125eb54ac0ef3600ed35cb97f599e":{"kind":"REWORK","_number":4,"created":"2021-11-19 20:07:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/07/817507/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/07/817507/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/817507/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/817507/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/817507/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/07/817507/4"}}},"commit":{"parents":[{"commit":"edf88041b3ff51209289fd0d1bb7c77e28c5da68","subject":"Merge \"Retire puppet-senlin - Step 3: Remove Project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/edf88041b3ff51209289fd0d1bb7c77e28c5da68"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-11 01:36:06.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-19 20:06:22.000000000","tz":-360},"subject":"Retire training-labs: remove project infra","message":"Retire training-labs: remove project infra\n\njob \u0027publish-training-labs-scripts\u0027 definition is not\nremoved as stable branch use this job. We will clean this\nin followup patches.\n\nDepends-On: https://review.opendev.org/c/openstack/training-labs/+/817505\nChange-Id: I3394c227fbbc07f1e797da7274781d0bc086e7ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a77159060a125eb54ac0ef3600ed35cb97f599e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a77159060a125eb54ac0ef3600ed35cb97f599e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
