)]}'
{"id":"openstack%2Fproject-config~818356","triplet_id":"openstack%2Fproject-config~master~I536dddec30e94875dfc13414da484802432e3f7a","project":"openstack/project-config","branch":"master","topic":"fix-acls-yoga-release","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2021-11-20 17:31:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I536dddec30e94875dfc13414da484802432e3f7a","subject":"Fix ACLs for deliverables under release management","status":"MERGED","created":"2021-11-18 09:51:01.000000000","updated":"2021-11-20 17:59:09.000000000","submitted":"2021-11-20 17:31:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818356-fix-acls-yoga-release","meta_rev_id":"276af6146cadfc8437b6d6498fe9aaee3ea70764","_number":818356,"virtual_id_number":818356,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-20 17:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2021-11-20 17:22:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2021-11-18 12:59:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2021-11-19 15:25:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2021-11-19 14:12:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2021-11-20 17:22:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-18 09:59:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-18 12:59:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-11-19 14:12:20.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-19 15:25:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-11-20 17:22:05.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"1dbd3aa6f70214075eb6bd6b81fcdfbf69e0c603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-11-18 09:51:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0994617ce0ff4d32c2ba4373fbad7923d932a5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 09:59:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dede569b18574bc79f7957d0811046c4 : SUCCESS in 6m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4ebf4b88647e44e8b1c37e063ecbaffd : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f2b28579299796953f4a77f00042a73ded3d21db","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-18 12:59:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2421b1e13eee75401912553e06b076a1f265fc7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-19 14:12:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9d738f7ab01838583384d2c0b78ffd4381908eb8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-19 15:25:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fb35a4927b2b169451ef110a162ab7b325a11486","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-11-20 17:22:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95efd3ce754a844ea6ca0bec26740ccafe692793","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 17:22:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e96c4a796dd9b12a1a8a2710d55bd43cca3ec66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 17:31:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ed178cfa5d3d40d8bdfffe738389db6f : SUCCESS in 5m 11s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/719e0fbef2fd4836a141e3a3683b875d : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"61aec39d4416d2656dc692245d29304dee30e98e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 17:31:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"276af6146cadfc8437b6d6498fe9aaee3ea70764","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 17:59:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/91558b3d614e43a7a24e3b87b9bff96b : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64574c3c212a47376fdbb98bcd9ff5558fcb18cf","revisions":{"64574c3c212a47376fdbb98bcd9ff5558fcb18cf":{"kind":"REWORK","_number":1,"created":"2021-11-18 09:51:01.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/56/818356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/818356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/818356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/818356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/818356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/818356/1"}}},"commit":{"parents":[{"commit":"d7f83941154003d2cde324182f1530461ca3c28c","subject":"Merge \"Set Fedora 34 min-ready to zero\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7f83941154003d2cde324182f1530461ca3c28c"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2021-11-18 09:47:58.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2021-11-18 09:47:58.000000000","tz":60},"subject":"Fix ACLs for deliverables under release management","message":"Fix ACLs for deliverables under release management\n\nYoga added tap-as-a-service as an OpenStack deliverable,\nadjust ACLs so that it goes through openstack/releases\nfor any future release.\n\nChange-Id: I536dddec30e94875dfc13414da484802432e3f7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/64574c3c212a47376fdbb98bcd9ff5558fcb18cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/64574c3c212a47376fdbb98bcd9ff5558fcb18cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
