)]}'
{"id":"openstack%2Fproject-config~818723","triplet_id":"openstack%2Fproject-config~master~I05312e296eed2854c5fc45c94038d570028d08f9","project":"openstack/project-config","branch":"master","topic":"add_openeuler","attention_set":{"15054":{"account":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"last_update":"2022-03-16 03:36:45.000000000","reason":"xinliang replied on the change"}},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2021-12-03 07:47:26.000000000","reason":"Ian Wienand replied on the change"}},"hashtags":[],"change_id":"I05312e296eed2854c5fc45c94038d570028d08f9","subject":"Add openEuler 20.03 LTS SP2 node","status":"MERGED","created":"2021-11-22 09:10:32.000000000","updated":"2022-03-16 03:36:45.000000000","submitted":"2021-12-14 04:56:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"818723-add_openeuler","meta_rev_id":"00e45fff794f46e27aa4242360b6b491791263a6","_number":818723,"virtual_id_number":818723,"owner":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-14 04:56:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-12-13 23:08:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-03 07:54:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-14 04:48:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-22 09:20:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-23 07:39:20.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2021-12-13 23:08:17.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-03-16 03:36:45.000000000","updated_by":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"reviewer":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"state":"CC"}],"messages":[{"id":"1f18e9c22cc55d8bba9365c34d1924e699ede992","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2021-11-22 09:10:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26cca9c8adfc13957637584426fdfe6940459efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 09:20:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e6c48e5ea32a46be80af0bf1cd4768f3 : SUCCESS in 5m 41s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b917161e518f476bb8922bcd05d84f7c : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a5be6b167b08b04c69479500fa05f80b71e555fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2021-11-23 07:08:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d97cf533b29032ab29bae77cf97381734c984f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 07:19:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/464711bbaf634724a1892b468d3762c9 : SUCCESS in 5m 38s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b0bb9e2bd6c04f53864612f460e69f28 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"281c29903d171dacb91199d76ac161452f1e7665","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2021-11-23 07:39:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f35ca4dabba5aee0c2f5596e26aa68c6a62c75c6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-12-03 03:32:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"91c144e327c310923a225569e73f10ef88fe4966","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2021-12-03 03:44:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bfc68b046b8824ee67c301c056b09f72e93153c0","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2021-12-03 03:45:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"47852e9ba6bc0db0801ca46cdcb639d277807c2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 03:52:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7cae4ee169124e67a0972733f7c4beed : SUCCESS in 5m 30s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/452c5483d91e4d6baa9399e07e0e9e3d : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":3},{"id":"b35b59cabf24b29db9e1d966f84a4c3c95a74d55","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-12-03 07:47:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b8ec2b92c1159ba4e4d9e5633bef42b5abe1fe1c","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-12-03 07:54:18.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bd5b4cb6dd3713fd96b52598d8a5b94119584e5c","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2021-12-09 01:34:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4985de7ef141ee7e195c83ef6f6751db51bf301a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-12-13 23:08:17.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"63afb76959760d49ea90e371c05627e006d2abc6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-12-14 04:48:22.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c6deeed3e4d50d1e03d215e59f98d26138b9d22d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 04:48:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d586613d0f2cf61a5aa13ad7149d10cc3f0bf378","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 04:56:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a854694e44e045019d174536855e0ce5 : SUCCESS in 6m 22s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/a8f76ecad0404624bf96d0ec45dcb6ce : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":3},{"id":"6a2f2731c1b752c15cd609b013e91133d487997c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 04:56:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"fadd9c04913a2348a58cee7e27ceb90d7994bc9d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 06:46:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/84a1c504cc224aa1a37b5ba8242d8127 : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":3},{"id":"00e45fff794f46e27aa4242360b6b491791263a6","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2022-03-16 03:36:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d55da5dac03429bc0372b63a33dd3c7765fad5c4","revisions":{"5ece5481c8ff669a57cae949d0f3a812a791f604":{"kind":"REWORK","_number":1,"created":"2021-11-22 09:10:32.000000000","uploader":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"ref":"refs/changes/23/818723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/818723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/818723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/818723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/818723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/818723/1"}}},"commit":{"parents":[{"commit":"0e7b903216e5721d5bee6b5ffd79052f8306c29f","subject":"Merge \"Remove \u0027publish-training-labs-scripts\u0027 definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0e7b903216e5721d5bee6b5ffd79052f8306c29f"}]}],"author":{"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","date":"2021-11-22 09:02:46.000000000","tz":480},"committer":{"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","date":"2021-11-22 09:10:13.000000000","tz":480},"subject":"Add openEuler 20.03 LTS SP2 node","message":"Add openEuler 20.03 LTS SP2 node\n\nopenEuler diskimage element has been added[1]. This Patch plans\nto add it to nodepool in order to support openeuler nodes in\nOpenStack CI system.\n\n[1]: https://review.opendev.org/c/openstack/diskimage-builder/+/784363\n\nChange-Id: I05312e296eed2854c5fc45c94038d570028d08f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ece5481c8ff669a57cae949d0f3a812a791f604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ece5481c8ff669a57cae949d0f3a812a791f604"}]},"branch":"refs/heads/master"},"b40b40fbd3e6879accf7d58a943d96f687df696b":{"kind":"REWORK","_number":2,"created":"2021-11-23 07:08:18.000000000","uploader":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"ref":"refs/changes/23/818723/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/818723/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/818723/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/818723/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/818723/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/818723/2"}}},"commit":{"parents":[{"commit":"0e7b903216e5721d5bee6b5ffd79052f8306c29f","subject":"Merge \"Remove \u0027publish-training-labs-scripts\u0027 definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0e7b903216e5721d5bee6b5ffd79052f8306c29f"}]}],"author":{"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","date":"2021-11-22 09:02:46.000000000","tz":480},"committer":{"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","date":"2021-11-23 07:06:17.000000000","tz":480},"subject":"Add openEuler 20.03 LTS SP2 node","message":"Add openEuler 20.03 LTS SP2 node\n\nopenEuler diskimage element has been added[1]. This Patch plans\nto build it in nodepool in order to support openEuler nodes in\nOpenStack CI system.\n\nThis kind of nodes will be used as devstack CI job at least[2].\n\n[1]: https://review.opendev.org/c/openstack/diskimage-builder/+/784363\n[2]: https://review.opendev.org/c/openstack/devstack/+/760790\n\nChange-Id: I05312e296eed2854c5fc45c94038d570028d08f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b40b40fbd3e6879accf7d58a943d96f687df696b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b40b40fbd3e6879accf7d58a943d96f687df696b"}]},"branch":"refs/heads/master"},"d55da5dac03429bc0372b63a33dd3c7765fad5c4":{"kind":"REWORK","_number":3,"created":"2021-12-03 03:44:41.000000000","uploader":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"ref":"refs/changes/23/818723/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/818723/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/818723/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/818723/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/818723/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/818723/3"}}},"commit":{"parents":[{"commit":"0e7b903216e5721d5bee6b5ffd79052f8306c29f","subject":"Merge \"Remove \u0027publish-training-labs-scripts\u0027 definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0e7b903216e5721d5bee6b5ffd79052f8306c29f"}]}],"author":{"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","date":"2021-11-22 09:02:46.000000000","tz":480},"committer":{"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","date":"2021-12-03 03:44:18.000000000","tz":480},"subject":"Add openEuler 20.03 LTS SP2 node","message":"Add openEuler 20.03 LTS SP2 node\n\nopenEuler diskimage element has been added[1]. This Patch plans\nto build it in nodepool in order to support openEuler nodes in\nOpenStack CI system.\n\nThis kind of nodes will be used as devstack CI job at least[2].\n\n[1]: https://review.opendev.org/c/openstack/diskimage-builder/+/784363\n[2]: https://review.opendev.org/c/openstack/devstack/+/760790\n\nChange-Id: I05312e296eed2854c5fc45c94038d570028d08f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d55da5dac03429bc0372b63a33dd3c7765fad5c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d55da5dac03429bc0372b63a33dd3c7765fad5c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
