)]}'
{"id":"openstack%2Fproject-config~820473","triplet_id":"openstack%2Fproject-config~master~If224a44eb3a0727cdb34c7f016c6407bac610549","project":"openstack/project-config","branch":"master","topic":"iweb","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2021-12-06 15:49:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If224a44eb3a0727cdb34c7f016c6407bac610549","subject":"iweb: disable image uploads","status":"MERGED","created":"2021-12-06 04:12:17.000000000","updated":"2021-12-06 15:57:03.000000000","submitted":"2021-12-06 15:49:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820473-iweb","meta_rev_id":"a62de243436b6833b00fbaca84cad08be2d83a3c","_number":820473,"virtual_id_number":820473,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-06 15:49:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-06 15:26:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-06 15:26:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-06 04:23:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-06 15:23:25.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"0f08273ddbc4b5bd73324d2b9d4d81c6edc3552f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-12-06 04:12:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fd9f36bd801dc3e294f3188ce09300fd6b62217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 04:23:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c5ea5a611ddd4b9e9e24bc78a5e4d76d : SUCCESS in 5m 57s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b70825d3bd554087b3d1a00776add932 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"aa4de8117fc12b8f20d083c030feaed1e24b9d35","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-12-06 15:23:25.000000000","message":"Patch Set 1: Code-Review-1\n\nThis is for one of the launchers, not the builders? We seem to need to pause the images in the builder config (nodepool.yaml) instead.","accounts_in_message":[],"_revision_number":1},{"id":"e6ceac36d2a1846c9823b6c10acb2a621dac5aa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-12-06 15:26:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4fb8c380bfa51b91134a186f8ace9fc11fdd340f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-12-06 15:26:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nEmergency-approving and manually applying to the builders in the interim.","accounts_in_message":[],"_revision_number":2},{"id":"cfd1e68ac8d98677dc533f2084e635adfa8180b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 15:39:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1bb66870936d46fca3ac0aef7158f0ee : SUCCESS in 6m 31s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/e4dc8153417f4bff90aa1f890bcd2837 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":2},{"id":"4d16d67adbf2e13dffadba42dece154a556ee47c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 15:40:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8e3311dc8ff189e1c8e1d561a119498a99d0811b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 15:49:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9af79b3f8f764ad2b4157e72f9ce2702 : SUCCESS in 7m 57s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/461ed0bb78af4ca9b1cbd0a959687baf : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":2},{"id":"a27460a95706c651e063b1d4d7211a35e5db34fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 15:49:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a62de243436b6833b00fbaca84cad08be2d83a3c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 15:57:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/676a19bc2750495ba29cf3ccb437706c : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1580f438532477750ec3c9f21a175319c04a29e4","revisions":{"930a66d6e799963477270d5361b974a4bd1a8a81":{"kind":"REWORK","_number":1,"created":"2021-12-06 04:12:17.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/73/820473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/820473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/820473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/820473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/820473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/820473/1"}}},"commit":{"parents":[{"commit":"000aadc414c8570c443163dd1621d52e2d4c4fda","subject":"Merge \"Remove report-build-page from zuul tenant config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/000aadc414c8570c443163dd1621d52e2d4c4fda"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-12-06 04:10:44.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-12-06 04:10:44.000000000","tz":660},"subject":"iweb: disable image uploads","message":"iweb: disable image uploads\n\nWe are having issues talking to https://image.api.mtl01.cloud.iweb.com\nas it appears to be missing an intermediate certificate.\n\nThis has been reported to the provider (Ticket #4526047)\n\nIn the mean time, disable uploads, which are in a failure loop.\n\nChange-Id: If224a44eb3a0727cdb34c7f016c6407bac610549\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/930a66d6e799963477270d5361b974a4bd1a8a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/930a66d6e799963477270d5361b974a4bd1a8a81"}]},"branch":"refs/heads/master"},"1580f438532477750ec3c9f21a175319c04a29e4":{"kind":"REWORK","_number":2,"created":"2021-12-06 15:26:02.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/73/820473/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/820473/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/820473/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/820473/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/820473/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/820473/2"}}},"commit":{"parents":[{"commit":"000aadc414c8570c443163dd1621d52e2d4c4fda","subject":"Merge \"Remove report-build-page from zuul tenant config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/000aadc414c8570c443163dd1621d52e2d4c4fda"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-12-06 04:10:44.000000000","tz":660},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-12-06 15:25:54.000000000","tz":0},"subject":"iweb: disable image uploads","message":"iweb: disable image uploads\n\nWe are having issues talking to https://image.api.mtl01.cloud.iweb.com\nas it appears to be missing an intermediate certificate.\n\nThis has been reported to the provider (Ticket #4526047)\n\nIn the mean time, disable uploads, which are in a failure loop.\n\nChange-Id: If224a44eb3a0727cdb34c7f016c6407bac610549\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1580f438532477750ec3c9f21a175319c04a29e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1580f438532477750ec3c9f21a175319c04a29e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
