)]}'
{"id":"openstack%2Fproject-config~822222","triplet_id":"openstack%2Fproject-config~master~Ic690a639992446cc4d39d4ce343ab2d725321d64","project":"openstack/project-config","branch":"master","topic":"project-rename","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-01-24 22:13:50.000000000","reason":"Change was submitted"},"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2022-01-24 22:13:50.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-01-24 22:13:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic690a639992446cc4d39d4ce343ab2d725321d64","subject":"Rename skyline/skyline* project to openstack/skyline*","status":"MERGED","created":"2021-12-18 11:30:52.000000000","updated":"2022-01-24 22:33:50.000000000","submitted":"2022-01-24 22:13:50.000000000","submitter":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"822222-project-rename","meta_rev_id":"1f4c8d38bfa58223c93d555f50ca8dc47dc3cf97","_number":822222,"virtual_id_number":822222,"owner":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-01-24 22:13:50.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-20 21:37:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-01-20 22:25:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2022-01-21 09:22:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-01-24 22:13:50.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2022-01-24 19:57:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2022-01-24 22:13:50.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}],"CC":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-18 11:31:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-18 11:37:52.000000000","updated_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"CC"},{"updated":"2021-12-18 11:37:52.000000000","updated_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2021-12-18 11:37:52.000000000","updated_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-01-20 22:25:44.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-01-21 09:22:10.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-01-24 19:57:57.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-01-24 22:13:50.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"state":"REVIEWER"}],"messages":[{"id":"0472c045822b197a54776727c4029d0db271bdac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2021-12-18 11:30:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51e3e94d9eb656f57a30f22c4e558c850890a510","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-18 11:31:17.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project \"openstack/skyline-apiserver\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/skyline-apiserver\n      templates:\n        - official-openstack-repo-jobs\n        - publish-to-pypi\n      check:\n        queue: skyline\n      gate:\n        queue: skyline\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 5446, column 3","accounts_in_message":[],"_revision_number":1},{"id":"67f8f9e5a790a6815a47325b69f93dbcf130159b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2021-12-18 11:39:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3b3ec02a1a2a6cbb65cdb168ca5d9428de7107a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-18 11:39:47.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project \"openstack/skyline-apiserver\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/skyline-apiserver\n      templates:\n        - official-openstack-repo-jobs\n        - publish-to-pypi\n      check:\n        queue: skyline\n      gate:\n        queue: skyline\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 5446, column 3","accounts_in_message":[],"_revision_number":2},{"id":"35a561d3652886b04fdca754a541076a3760cad8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-12-19 18:27:04.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"107414bc40ad173346b0a3a3e68702826e3fe60d","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2021-12-20 04:42:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2273b3bc208df5660b250648d08d68dd470e9bea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-01-20 21:26:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e984c023b2e3c9ee4e563da3291241c3e4cb9a64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 21:37:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/907e585c4dc44890b2f967757cbfc37e : SUCCESS in 9m 08s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8e9a75953c8748ceb3638a0e4aa940db : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":3},{"id":"29a5dcf06ebb126c62eebe11e8cc612a7c7849e5","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-01-20 22:25:44.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7cbb99d726e55f9dd0711b7fddf98f5a8558552e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-01-20 22:26:31.000000000","message":"Topic set to project-rename","accounts_in_message":[],"_revision_number":3},{"id":"1c3c5c44accb0b507cd3c8eb1da38acd22c33d9d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-01-21 09:22:10.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3d40d93568a54d828dba94ba8c79bc6d41b3cf1f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-24 19:57:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e140e80c416538b615b15caefe54879609099c9b","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-01-24 22:13:50.000000000","message":"Patch Set 3: Verified+2 Workflow+1 Code-Review+2\n\nBypassing Zuul to merge this.","accounts_in_message":[],"_revision_number":3},{"id":"279dfb7ea424a91991439481296384ea5eaabf08","tag":"autogenerated:gerrit:merged","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-01-24 22:13:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1f4c8d38bfa58223c93d555f50ca8dc47dc3cf97","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 22:33:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/d1c99de5fff54552b9ccf69a2101e824 : SUCCESS in 5m 24s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/5f56e4324697427e97962e68440dc436 : SUCCESS in 8m 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f5fade9441057d099f4aa62b739b174b0d82427e","revisions":{"2b3d51bc490770e73c808b4e34124c6f06ec9bfb":{"kind":"REWORK","_number":1,"created":"2021-12-18 11:30:52.000000000","uploader":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"ref":"refs/changes/22/822222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/822222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/822222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/822222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/822222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/822222/1"}}},"commit":{"parents":[{"commit":"4a058ac8bf4f7b23a37bd0d4585ae12af71561d7","subject":"Add cache-devstack element for openEuler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a058ac8bf4f7b23a37bd0d4585ae12af71561d7"}]}],"author":{"name":"Wu Wenxiang","email":"wu.wenxiang@99cloud.net","date":"2021-12-18 11:20:06.000000000","tz":480},"committer":{"name":"Wu Wenxiang","email":"wu.wenxiang@99cloud.net","date":"2021-12-18 11:30:47.000000000","tz":480},"subject":"Rename skyline/skyline* project to openstack/skyline*","message":"Rename skyline/skyline* project to openstack/skyline*\n\nSkyline project is officially part of OpenStack\ngovernance now[1]. As agreed by TC, repo rename needs\nto be done after project is approved by TC (which is done\nnow).\n\nAs next step, this commit renames all the Skyline projects repo\nfrom \u0027skyline/\u0027 \u0027openstack/\u0027 namespace.\n\nRepo renamed are:\n- openstack/skyline-apiserver\n- openstack/skyline-console\n\n[1] https://review.opendev.org/c/openstack/governance/+/814037\n\nChange-Id: Ic690a639992446cc4d39d4ce343ab2d725321d64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2b3d51bc490770e73c808b4e34124c6f06ec9bfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2b3d51bc490770e73c808b4e34124c6f06ec9bfb"}]},"branch":"refs/heads/master"},"e863cbac284afca960447e3bfbfabc7d16f51256":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-18 11:39:25.000000000","uploader":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"ref":"refs/changes/22/822222/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/822222/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/822222/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/822222/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/822222/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/822222/2"}}},"commit":{"parents":[{"commit":"4a058ac8bf4f7b23a37bd0d4585ae12af71561d7","subject":"Add cache-devstack element for openEuler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a058ac8bf4f7b23a37bd0d4585ae12af71561d7"}]}],"author":{"name":"Wu Wenxiang","email":"wu.wenxiang@99cloud.net","date":"2021-12-18 11:20:06.000000000","tz":480},"committer":{"name":"Wu Wenxiang","email":"wu.wenxiang@99cloud.net","date":"2021-12-18 11:39:06.000000000","tz":480},"subject":"Rename skyline/skyline* project to openstack/skyline*","message":"Rename skyline/skyline* project to openstack/skyline*\n\nSkyline project is officially part of OpenStack\ngovernance now[1]. As agreed by TC, repo rename needs\nto be done after project is approved by TC (which is done\nnow).\n\nAs next step, this commit renames all the Skyline projects repo\nfrom \u0027skyline/\u0027 to \u0027openstack/\u0027 namespace.\n\nRepo renamed are:\n- openstack/skyline-apiserver\n- openstack/skyline-console\n\n[1] https://review.opendev.org/c/openstack/governance/+/814037\n\nChange-Id: Ic690a639992446cc4d39d4ce343ab2d725321d64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e863cbac284afca960447e3bfbfabc7d16f51256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e863cbac284afca960447e3bfbfabc7d16f51256"}]},"branch":"refs/heads/master"},"f5fade9441057d099f4aa62b739b174b0d82427e":{"kind":"REWORK","_number":3,"created":"2022-01-20 21:26:06.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/22/822222/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/822222/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/822222/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/822222/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/822222/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/822222/3"}}},"commit":{"parents":[{"commit":"4a058ac8bf4f7b23a37bd0d4585ae12af71561d7","subject":"Add cache-devstack element for openEuler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a058ac8bf4f7b23a37bd0d4585ae12af71561d7"}]}],"author":{"name":"Wu Wenxiang","email":"wu.wenxiang@99cloud.net","date":"2021-12-18 11:20:06.000000000","tz":480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-01-20 21:24:10.000000000","tz":-480},"subject":"Rename skyline/skyline* project to openstack/skyline*","message":"Rename skyline/skyline* project to openstack/skyline*\n\nSkyline project is officially part of OpenStack\ngovernance now[1]. As agreed by TC, repo rename needs\nto be done after project is approved by TC (which is done\nnow).\n\nAs next step, this commit renames all the Skyline projects repo\nfrom \u0027skyline/\u0027 to \u0027openstack/\u0027 namespace.\n\nRepo renamed are:\n- openstack/skyline-apiserver\n- openstack/skyline-console\n\n[1] https://review.opendev.org/c/openstack/governance/+/814037\n\nChange-Id: Ic690a639992446cc4d39d4ce343ab2d725321d64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f5fade9441057d099f4aa62b739b174b0d82427e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f5fade9441057d099f4aa62b739b174b0d82427e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
