)]}'
{"id":"openstack%2Fproject-config~824229","triplet_id":"openstack%2Fproject-config~master~I22f590307653e91bc7742dbf2feffc03a981e10c","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-01-25 14:06:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I22f590307653e91bc7742dbf2feffc03a981e10c","subject":"Add Backport-Candidate label to openstack-ansible ACL","status":"MERGED","created":"2022-01-11 20:07:33.000000000","updated":"2022-01-25 14:46:58.000000000","submitted":"2022-01-25 14:06:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824229","meta_rev_id":"0e8d86b0990c3034444b60599ad11c06d84a5400","_number":824229,"virtual_id_number":824229,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 14:06:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2022-01-12 16:19:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-25 13:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-25 13:57:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-11 20:25:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-12 16:19:51.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2022-01-25 13:57:22.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"20747a27df51af60708442552d9bf3697b15cb07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-11 20:07:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa4d3c9e5f667633e9dcffad3dc08a931a381cda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-11 20:11:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"73657889984852dfb2196cbbd023994a9c24cc38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 20:25:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9df9014105d1487d9e931b942fbbe989 : SUCCESS in 8m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f08853390ffb4e9b9bc686ae46dd3ff8 : FAILURE in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"f8acd3c187a432b9e65754c4b5ec49532208ae83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-12 07:51:22.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"94a2bfd267fe7f5a2a9faf678a602d12dc578878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 08:01:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fb52fe3afb1641218d5d44450c3162d2 : SUCCESS in 7m 24s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/dfc8ac5e7a804fc888dbfdff6bf8965b : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":3},{"id":"a41fcbc2a6f6d0c4b298d8824619f2078e6de59b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-12 11:27:52.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"0cc17b4e5d85a4691690932c04033608c137fecb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 12:19:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ac4a25e6f1a647af91442c9920bdca22 : SUCCESS in 8m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f79dc82d809449f8a63a75a2fde9e77e : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":4},{"id":"86c8cbc00d1ce58e775d80e1c74809f651539392","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-01-12 16:19:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"125d75a4d0f6cc620b9e15cb1bee4cd154cdf279","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-25 13:57:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"96f59fa2e48ef50b66abb0d6cc58817cb6113509","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 13:57:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fe4cb1ac087cf8319d78699614e5c4ddab46cc7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 14:06:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b8cc4be2a31a4d62bd5ee401e2d6fc40 : SUCCESS in 7m 56s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/33a4fd1544d44b0ab419521de5fb8e0a : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":4},{"id":"81c26cdb9ae16e45e59410e424bf243f53b43f08","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 14:06:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"0e8d86b0990c3034444b60599ad11c06d84a5400","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 14:46:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/d2866d7daac448afa21a28f3ef4dd3fc : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e6008a2dd0e6b08139fd4f9f2186982ed208af47","revisions":{"769d4602548b71d7c34a0d4666c8629e6f3cf59b":{"kind":"REWORK","_number":1,"created":"2022-01-11 20:07:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/824229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/29/824229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/824229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/824229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/824229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/29/824229/1"}}},"commit":{"parents":[{"commit":"0a8cb7b85b2232e6887b333ca5d1abd8f71c8350","subject":"Merge \"Remove opensuse-tumbleweed from nodepool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a8cb7b85b2232e6887b333ca5d1abd8f71c8350"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-11 20:07:14.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-11 20:07:14.000000000","tz":120},"subject":"Add Backport-Candidate label to openstack-ansible ACL","message":"Add Backport-Candidate label to openstack-ansible ACL\n\nChange-Id: I22f590307653e91bc7742dbf2feffc03a981e10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/769d4602548b71d7c34a0d4666c8629e6f3cf59b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/769d4602548b71d7c34a0d4666c8629e6f3cf59b"}]},"branch":"refs/heads/master"},"689f6764e7dccc4a2bbc3520f3507aa79d80ce6c":{"kind":"REWORK","_number":2,"created":"2022-01-11 20:11:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/824229/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/29/824229/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/824229/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/824229/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/824229/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/29/824229/2"}}},"commit":{"parents":[{"commit":"0a8cb7b85b2232e6887b333ca5d1abd8f71c8350","subject":"Merge \"Remove opensuse-tumbleweed from nodepool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a8cb7b85b2232e6887b333ca5d1abd8f71c8350"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-11 20:07:14.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-11 20:11:59.000000000","tz":0},"subject":"Add Backport-Candidate label to openstack-ansible ACL","message":"Add Backport-Candidate label to openstack-ansible ACL\n\nChange-Id: I22f590307653e91bc7742dbf2feffc03a981e10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/689f6764e7dccc4a2bbc3520f3507aa79d80ce6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/689f6764e7dccc4a2bbc3520f3507aa79d80ce6c"}]},"branch":"refs/heads/master"},"9ac642c993e2da919a3cf5ad6d6988f4ae5ae913":{"kind":"REWORK","_number":3,"created":"2022-01-12 07:51:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/824229/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/29/824229/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/824229/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/824229/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/824229/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/29/824229/3"}}},"commit":{"parents":[{"commit":"0a8cb7b85b2232e6887b333ca5d1abd8f71c8350","subject":"Merge \"Remove opensuse-tumbleweed from nodepool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a8cb7b85b2232e6887b333ca5d1abd8f71c8350"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-11 20:07:14.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-12 07:51:22.000000000","tz":0},"subject":"Add Backport-Candidate label to openstack-ansible ACL","message":"Add Backport-Candidate label to openstack-ansible ACL\n\nChange-Id: I22f590307653e91bc7742dbf2feffc03a981e10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ac642c993e2da919a3cf5ad6d6988f4ae5ae913"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ac642c993e2da919a3cf5ad6d6988f4ae5ae913"}]},"branch":"refs/heads/master"},"e6008a2dd0e6b08139fd4f9f2186982ed208af47":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-01-12 11:27:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/824229/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/29/824229/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/824229/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/824229/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/824229/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/29/824229/4"}}},"commit":{"parents":[{"commit":"1906b1f21b6a9f97f84b66c2e16ee529fa08fc2c","subject":"Use same ACL for all OpenStack-Ansible Projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1906b1f21b6a9f97f84b66c2e16ee529fa08fc2c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-11 20:07:14.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-12 11:27:52.000000000","tz":0},"subject":"Add Backport-Candidate label to openstack-ansible ACL","message":"Add Backport-Candidate label to openstack-ansible ACL\n\nChange-Id: I22f590307653e91bc7742dbf2feffc03a981e10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e6008a2dd0e6b08139fd4f9f2186982ed208af47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e6008a2dd0e6b08139fd4f9f2186982ed208af47"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
