)]}'
{"id":"openstack%2Fproject-config~826305","triplet_id":"openstack%2Fproject-config~master~Iad8c1f83e247c9a8bcf5b4f530f7b83663e1f793","project":"openstack/project-config","branch":"master","topic":"release-acls","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-01-25 15:59:01.000000000","reason":"Clark Boylan replied on the change"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-01-25 16:33:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad8c1f83e247c9a8bcf5b4f530f7b83663e1f793","subject":"Grant Create Annotated Tag perms on bindep","status":"MERGED","created":"2022-01-25 15:10:35.000000000","updated":"2022-01-25 16:51:58.000000000","submitted":"2022-01-25 16:33:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826305-release-acls","meta_rev_id":"5c76ac373ff50b90e7fbbcc1c6eec70f50e136a0","_number":826305,"virtual_id_number":826305,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 16:33:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-01-25 15:59:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2022-01-25 15:59:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-25 15:12:52.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-01-25 15:23:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59ab15041b723c5693197b1f9ffd7a11a397d076","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-25 15:10:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ace1ba8a5d88a383d82822c908addb35d7b17a52","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-01-25 15:12:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5bbf07dc2d70602e43689861c953240fcdc4b3a3","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-01-25 15:13:27.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"29f3ade1ad7eac271f44b4d96d277fe937dbed90","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-25 15:18:37.000000000","message":"Set Work In Progress\n\nWe want to test Ia95919bcfe71ce488096584c784fe7376f66f34a first.","accounts_in_message":[],"_revision_number":1},{"id":"19a14bb046fe0035c77fb5d53a8544c8b188181c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 15:23:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3f28b0a60a234cc29bcdbfc0809c4178 : SUCCESS in 7m 10s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4803a62b1c524e4f9b3f54c6182948ab : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3f7e52beaae68dae167ef9751efcfb818693764b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-25 15:56:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4169ff51bc9e1efc997d9c134d009a6b0d25d11b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-25 15:57:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b3198ecfd6a1dc2dfad598c36178fbd5ad9605c0","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-25 15:58:28.000000000","message":"Set Ready For Review\n\nThe better option ended up not solving this.","accounts_in_message":[],"_revision_number":3},{"id":"73f9c79c7a0cc720f97f5e994d7ce8317f1c5dcf","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-01-25 15:59:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5bdcc83f3414901f38af27ea021bede23c0f134d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:15:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/82ef9743b3044d33b3056016128ab1ca : SUCCESS in 11m 25s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/305d66c40b5a4637865621a9f12bcc4c : SUCCESS in 8m 11s","accounts_in_message":[],"_revision_number":3},{"id":"53ae8c80cb246c31cadc85b2d8e7849e268557ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:23:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"34b0b53466999a81d616d9a0e363329afe957893","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:33:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c48c14cea610495c831164297b3ec1c9 : SUCCESS in 8m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/207edc9a6b214f8780540628771f70cd : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":3},{"id":"147b35111fcdfa0d24f9d424340dff5ec361cdce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:33:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5c76ac373ff50b90e7fbbcc1c6eec70f50e136a0","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:51:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/36e9c328ee6e4faeba57550ca616e251 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"645ec7b07fab6eec6ae29fff564920fc3580ae61","revisions":{"e210b7aeaabf710ab4b5388789475e00078ca40e":{"kind":"REWORK","_number":1,"created":"2022-01-25 15:10:35.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/05/826305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/05/826305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/826305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/826305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/826305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/05/826305/1"}}},"commit":{"parents":[{"commit":"da6e5fd46c936a2ab026009468ef014d5a3ec280","subject":"Merge \"Add Backport-Candidate label to openstack-ansible ACL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/da6e5fd46c936a2ab026009468ef014d5a3ec280"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-25 15:07:10.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-25 15:10:30.000000000","tz":0},"subject":"Grant Create Annotated Tag perms on bindep","message":"Grant Create Annotated Tag perms on bindep\n\nWe\u0027re testing a potential workaround for a suspected regression in\nGerrit 3.4, where signed tags are rejected with the error \"You need\n\u0027Create Tag\u0027 rights to push a normal tag.\" Temporarily grant this\nfor the opendev/bindep project, so we can see if it works around the\nproblem while we coordinate a fix with Gerrit upstream.\n\nChange-Id: Iad8c1f83e247c9a8bcf5b4f530f7b83663e1f793\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e210b7aeaabf710ab4b5388789475e00078ca40e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e210b7aeaabf710ab4b5388789475e00078ca40e"}]},"branch":"refs/heads/master"},"62066d516b9e83efd5112e00e843a9cc9af891b1":{"kind":"REWORK","_number":2,"created":"2022-01-25 15:56:14.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/05/826305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/05/826305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/826305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/826305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/826305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/05/826305/2"}}},"commit":{"parents":[{"commit":"7e5aaec79267ef563994ad0f3005406751157784","subject":"Update bindep ACL to use new createSignedTag perm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e5aaec79267ef563994ad0f3005406751157784"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-25 15:07:10.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-25 15:55:08.000000000","tz":0},"subject":"Grant Create Annotated Tag perms on bindep","message":"Grant Create Annotated Tag perms on bindep\n\nWe\u0027re testing a potential workaround for a suspected regression in\nGerrit 3.4, where signed tags are rejected with the error \"You need\n\u0027Create Tag\u0027 rights to push a normal tag.\" Temporarily grant this\nfor the opendev/bindep project, so we can see if it works around the\nproblem while we coordinate a fix with Gerrit upstream.\n\nChange-Id: Iad8c1f83e247c9a8bcf5b4f530f7b83663e1f793\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62066d516b9e83efd5112e00e843a9cc9af891b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62066d516b9e83efd5112e00e843a9cc9af891b1"}]},"branch":"refs/heads/master"},"645ec7b07fab6eec6ae29fff564920fc3580ae61":{"kind":"REWORK","_number":3,"created":"2022-01-25 15:57:47.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/05/826305/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/05/826305/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/826305/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/826305/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/826305/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/05/826305/3"}}},"commit":{"parents":[{"commit":"7e5aaec79267ef563994ad0f3005406751157784","subject":"Update bindep ACL to use new createSignedTag perm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e5aaec79267ef563994ad0f3005406751157784"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-25 15:07:10.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-25 15:57:35.000000000","tz":0},"subject":"Grant Create Annotated Tag perms on bindep","message":"Grant Create Annotated Tag perms on bindep\n\nWe\u0027re testing a potential workaround for a suspected regression in\nGerrit 3.4, where signed tags are rejected with the error \"You need\n\u0027Create Tag\u0027 rights to push a normal tag.\" Temporarily grant this\nfor the opendev/bindep project, so we can see if it works around the\nproblem while we coordinate a fix with Gerrit upstream.\n\nChange-Id: Iad8c1f83e247c9a8bcf5b4f530f7b83663e1f793\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/645ec7b07fab6eec6ae29fff564920fc3580ae61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/645ec7b07fab6eec6ae29fff564920fc3580ae61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
