)]}'
{"id":"openstack%2Fproject-config~826334","triplet_id":"openstack%2Fproject-config~master~Ia426ea36b4e6877fdce5725ff1e00ae02c62e3f4","project":"openstack/project-config","branch":"master","topic":"release-acls","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-01-25 18:30:04.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ia426ea36b4e6877fdce5725ff1e00ae02c62e3f4","subject":"Work around signed tag regression from Gerrit 3.4","status":"MERGED","created":"2022-01-25 17:15:48.000000000","updated":"2022-01-25 19:18:09.000000000","submitted":"2022-01-25 18:58:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826334-release-acls","meta_rev_id":"3158dcea8fd63510b760b5804e36f18341ceef48","_number":826334,"virtual_id_number":826334,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 18:58:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2022-01-25 18:20:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2022-01-25 18:12:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-25 18:44:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-25 17:31:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 18:12:58.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-01-25 18:20:01.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"}],"messages":[{"id":"f6e111148f75ca26757256cdb010c31116bf80e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-25 17:15:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a81539bc8a285193e72959f4cc3c4b4f4287c4d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 17:31:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4bcc4c8b2964493692fa64e4a18e40af : SUCCESS in 7m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f40357f8ee754094979e4451f109ec0e : FAILURE in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d0a71882138b53be526f0003134d65ea041ab1b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-25 17:43:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7c42223cbf1e3faf78057540a92fd70f58d23ecb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 17:57:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2999a173f2c9412db8f49d29d8dc48f5 : SUCCESS in 8m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a0586ee89dd044d0b9c8152486005e47 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"0662f3acbdb1130c8a3e46150d1a50695f2aa332","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-01-25 18:12:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2526996a7f23776eb3faf3cda57caa61993a6840","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2022-01-25 18:20:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"91227d7bec7075fb225c083e9559729a8530bb04","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-25 18:30:04.000000000","message":"Patch Set 2: Workflow+1\n\nSelf-approving this since it may take some time to deploy (and may need a manual manage-projects run to complete if Ansible times out). I\u0027ll send an announcement to the service-discuss ML once it\u0027s confirmed applied.","accounts_in_message":[],"_revision_number":2},{"id":"f64a702adcf46790094f9c25655af61e8eb6dda4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-25 18:30:27.000000000","message":"Patch Set 2: -Workflow\n\nEr, service-announce mailing list, I meant.","accounts_in_message":[],"_revision_number":2},{"id":"4107edd32238e58ab270a97b84523617d89f7111","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 18:31:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6ba6c61374811092d5bc06e6f1f11f2f43ffe8e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 18:42:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/18c76e4409744eb1bb608663cfa52a8f : SUCCESS in 7m 24s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/51b29c1adb4f462493b70bbe9adbe120 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"2bd1a9b8ddb9fecf75c99ce14165305b8662884e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-25 18:44:42.000000000","message":"Patch Set 2: Workflow+1\n\n*sigh* My follow-up comment cleared my workflow vote and I didn\u0027t notice.","accounts_in_message":[],"_revision_number":2},{"id":"15d2ea39d629cdf9b74816c56c72c34eff85e519","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 18:45:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ac1a1b0e15e2377557d460b97bf87b910c48ff5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-25 18:57:44.000000000","message":"Patch Set 2:\n\nThe updated Infra Manual examples are in change Ie7cca062844101a685befe3b4f5d868b9506c5c0 now.","accounts_in_message":[],"_revision_number":2},{"id":"e771e35791a388f36b18c3764e04fb0b871e784c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 18:58:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/37fb545f8f70409e8040682e55e81ab1 : SUCCESS in 8m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7eac980816174204b8fb0e4a44c8fcb2 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"708bfc7439ce8ebae84d02bb682b9423551ca95a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 18:58:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3158dcea8fd63510b760b5804e36f18341ceef48","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 19:18:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/64737ccff60d48d5a42956ecc9175f3c : SUCCESS in 18m 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"83ca7a97f9ddcb65fb82bb16247b258458208907","revisions":{"a65bf013a489fe591813ea926cef709ae7d6fd99":{"kind":"REWORK","_number":1,"created":"2022-01-25 17:15:48.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/34/826334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/826334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/826334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/826334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/826334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/826334/1"}}},"commit":{"parents":[{"commit":"645ec7b07fab6eec6ae29fff564920fc3580ae61","subject":"Grant Create Annotated Tag perms on bindep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/645ec7b07fab6eec6ae29fff564920fc3580ae61"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-25 17:06:12.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-25 17:15:00.000000000","tz":0},"subject":"Work around signed tag regression from Gerrit 3.4","message":"Work around signed tag regression from Gerrit 3.4\n\nUpon upgrading from Gerrit 3.3 to 3.4, a regression was observed in\nwhich jgit no longer returns signatures in its tag messages, causing\nGerrit to misidentify signed tags as unsigned (annotated) tags.\nBecause our ACLs only allow signed tags to be pushed, this\nregression prevents Gerrit from accepting them now.\n\nTemporarily grant permission to push unsigned tags to anyone who\nhas permission to push signed ones. We will revert that as soon as a\nfixed Gerrit is in place, but in the meantime users will be warned\nto take care when pushing tags so that they don\u0027t accidentally push\nactually unsigned tags to Gerrit.\n\nAlso, the pushSignedTag keyword was deprecated in favor of the new\ncreateSignedTag name, so go ahead and update to that while we\u0027re\ndoing this so that we can limit the amount of churn across all these\nACLs. Documentation will be corrected to recommend the new format in\na separate change, but update the ACL linter now to prevent the old\nsyntax from being used in new projects.\n\nThis workaround was already tested on opendev/bindep in the parent\nIad8c1f83e247c9a8bcf5b4f530f7b83663e1f793 change, and confirmed to\nfunction as intended.\n\nChange-Id: Ia426ea36b4e6877fdce5725ff1e00ae02c62e3f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a65bf013a489fe591813ea926cef709ae7d6fd99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a65bf013a489fe591813ea926cef709ae7d6fd99"}]},"branch":"refs/heads/master"},"83ca7a97f9ddcb65fb82bb16247b258458208907":{"kind":"REWORK","_number":2,"created":"2022-01-25 17:43:11.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/34/826334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/826334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/826334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/826334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/826334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/826334/2"}}},"commit":{"parents":[{"commit":"645ec7b07fab6eec6ae29fff564920fc3580ae61","subject":"Grant Create Annotated Tag perms on bindep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/645ec7b07fab6eec6ae29fff564920fc3580ae61"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-25 17:06:12.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-25 17:40:31.000000000","tz":0},"subject":"Work around signed tag regression from Gerrit 3.4","message":"Work around signed tag regression from Gerrit 3.4\n\nUpon upgrading from Gerrit 3.3 to 3.4, a regression was observed in\nwhich jgit no longer returns signatures in its tag messages, causing\nGerrit to misidentify signed tags as unsigned (annotated) tags.\nBecause our ACLs only allow signed tags to be pushed, this\nregression prevents Gerrit from accepting them now.\n\nTemporarily grant permission to push unsigned tags to anyone who\nhas permission to push signed ones. We will revert that as soon as a\nfixed Gerrit is in place, but in the meantime users will be warned\nto take care when pushing tags so that they don\u0027t accidentally push\nactually unsigned tags to Gerrit.\n\nAlso, the pushSignedTag keyword was deprecated in favor of the new\ncreateSignedTag name, so go ahead and update to that while we\u0027re\ndoing this so that we can limit the amount of churn across all these\nACLs. Documentation will be corrected to recommend the new format in\na separate change, but update the ACL linter now to prevent the old\nsyntax from being used in new projects.\n\nThis workaround was already tested on opendev/bindep in the parent\nIad8c1f83e247c9a8bcf5b4f530f7b83663e1f793 change, and confirmed to\nfunction as intended.\n\nChange-Id: Ia426ea36b4e6877fdce5725ff1e00ae02c62e3f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/83ca7a97f9ddcb65fb82bb16247b258458208907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/83ca7a97f9ddcb65fb82bb16247b258458208907"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
