)]}'
{"id":"openstack%2Fproject-config~826969","triplet_id":"openstack%2Fproject-config~master~Ia863e313e656e67191e1ca3aea968a7f3e4059a6","project":"openstack/project-config","branch":"master","topic":"pip-22","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-01-30 17:50:16.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ia863e313e656e67191e1ca3aea968a7f3e4059a6","subject":"Make wheel indices valid HTML 5","status":"MERGED","created":"2022-01-30 16:39:00.000000000","updated":"2022-01-30 21:09:16.000000000","submitted":"2022-01-30 17:30:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826969-pip-22","meta_rev_id":"565010b4b95013489d50028a1f8d345bdec201d1","_number":826969,"virtual_id_number":826969,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-30 17:30:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-01-30 17:23:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2022-01-30 17:23:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-30 16:48:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-30 17:23:09.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-01-30 17:31:39.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"2359fbd15baed0091e4676527491b672ee9afea8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-30 16:39:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eef92993216c108cd43a4facfe255c51b7598d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-30 16:48:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2df8bc8be7914898af7572f4eeac67bb : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":1},{"id":"64c4f8e1b0bedc575f8ce6028bc20bb5f7ef21d0","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-01-30 17:23:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f77b36b7bc41662eab302a0c6e117f7b11d1ea2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-30 17:23:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6650f7fa95617c64bcc9e586e3c1895a7b0c79f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-30 17:30:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/80c2dc7b697d4f758809a2d222e7879f : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"247cb05695a030380220c2c624dce951ef1702b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-30 17:30:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3ea2e66a6580736b110b8044aeec49c228b8eb96","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-30 17:31:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4cdefd1e735f1c317d59487cc2264a0eed5b4987","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-30 17:50:16.000000000","message":"Patch Set 1:\n\nThe linked issue indicates that the check expected \"html\" strictly in lower-case, but that the maintainers are planning to hotfix it to also accept upper-case \"HTML\" there, so this might be enough as long as the check isn\u0027t sensitive about what follows it.","accounts_in_message":[],"_revision_number":1},{"id":"f935dba107728b724e5a58fff3f2713872acd409","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-30 17:55:28.000000000","message":"Patch Set 1:\n\nThe case insensitivity patch is https://github.com/pypa/pip/pull/10844 but I can\u0027t tell without more digging whether the decl passed into the HTMLLinkParser.handle_decl() method is trimmed or not. I have a feeling it expects the entire doctype element to match verbatim though, in which case you\u0027re right this is still a problem. I\u0027ve mentioned it as another broken situation in the main issue.","accounts_in_message":[],"_revision_number":1},{"id":"1899dfaa1555759eb0306a97ee00b85bf836daaa","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-30 18:03:03.000000000","message":"Patch Set 1:\n\nhttps://github.com/pypa/pip/pull/10847 is adding an automatic fallback to the old parser on errors, which should hopefully get us working again once new pip is released with it. In the meantime, we probably want to make sure our indices will work with the new parser (either by working out how to make mod_autoindex return something custom there, or making our own file list indices to include, or helping the pip maintainers improve their parser choice further).","accounts_in_message":[],"_revision_number":1},{"id":"ad2844607200e443d6aeac1c0ec2acce8d9476f1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-30 18:17:09.000000000","message":"Patch Set 1:\n\nIf we do need to fix it, we should be able to set SuppressHTMLPreamble in our IndexOptions for the mirror vhosts, and then provide a HeaderName directive pointing to a file which contains \"\u003c!DOCTYPE html\u003e\" instead.","accounts_in_message":[],"_revision_number":1},{"id":"d45c9c989c45a7b307fefb08516dfbff68746008","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-30 19:03:59.000000000","message":"Patch Set 1:\n\nI\u0027ve proposed that as a solution in Id71174954b13b80483256d37f773b781f4956c21 now.","accounts_in_message":[],"_revision_number":1},{"id":"565010b4b95013489d50028a1f8d345bdec201d1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-30 21:09:16.000000000","message":"Patch Set 1:\n\nI\u0027ve manually applied the difference from this change to the indices in AFS in advance of the periodic job runs, so we can hopefully get back on track sooner and spot any remaining fallout from pip 22.0.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c8b888054264d847f5ae55c456803f52ad0177b","revisions":{"7c8b888054264d847f5ae55c456803f52ad0177b":{"kind":"REWORK","_number":1,"created":"2022-01-30 16:39:00.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/69/826969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/826969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/826969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/826969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/826969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/826969/1"}}},"commit":{"parents":[{"commit":"f05cc72e98e8743518e3e715a02df4f7e5c38e71","subject":"Merge \"Add openstack-skyline irc channel in access an gerrit bot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f05cc72e98e8743518e3e715a02df4f7e5c38e71"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-30 16:37:01.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-30 16:37:01.000000000","tz":0},"subject":"Make wheel indices valid HTML 5","message":"Make wheel indices valid HTML 5\n\nPip 22.0 switched[*] its HTML parser library for one which demands\nstrict adherence to HTML 5, and so cannot work if told to include\nour wheel indices because they lack a doctype declaration on the\nfirst line. Add one.\n\n[*] https://github.com/pypa/pip/issues/10825\n\nChange-Id: Ia863e313e656e67191e1ca3aea968a7f3e4059a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7c8b888054264d847f5ae55c456803f52ad0177b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7c8b888054264d847f5ae55c456803f52ad0177b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
