)]}'
{"id":"openstack%2Fproject-config~827307","triplet_id":"openstack%2Fproject-config~master~I5e4db08f9d7600f7ea045907813d4fb0f6264436","project":"openstack/project-config","branch":"master","topic":"new-project","attention_set":{},"removed_from_attention_set":{"31289":{"account":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"last_update":"2022-02-07 12:16:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5e4db08f9d7600f7ea045907813d4fb0f6264436","subject":"Add dummy charm project","status":"ABANDONED","created":"2022-02-01 15:31:38.000000000","updated":"2022-02-07 12:16:49.000000000","total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"ac5232e5ba13abadc13a2fde5862283a19cb0fe7","_number":827307,"virtual_id_number":827307,"owner":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-01 16:04:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2022-02-03 07:44:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}],"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-01 16:04:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-02 21:39:27.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2022-02-03 07:44:27.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"40ed25dfeffeb1fa6a259089ef844d331a685bb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2022-02-01 15:31:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bcf893959c5a56b29fbf79f2c64662c9f2a7ffc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2022-02-01 15:47:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c1e458d7fd16da66bb58cd97294f83ee1f9178de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2022-02-01 15:50:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c9ba3b507fe4db93786de3cfbe42342e96004a88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 16:04:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ffc58a2fe69d42f3908c430a7f73f8c1 : SUCCESS in 8m 31s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/318ba716ff5549ec80247831a297691b : FAILURE in 4m 41s","accounts_in_message":[],"_revision_number":3},{"id":"452535dd35dd0e8ac7f731dab8b819135190791c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-02-02 21:39:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4983bb0c19d62360de53ef0d22d4a3f8c44d27c7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-02-03 07:44:27.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1fc14e5eba9c57a11e9ecbd805ebe52d39617e4c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-02-03 07:46:05.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ac5232e5ba13abadc13a2fde5862283a19cb0fe7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2022-02-07 12:16:49.000000000","message":"Abandoned\n\nThis is being validated by another project: https://review.opendev.org/c/openstack/project-config/+/827719/4/gerrit/projects.yaml","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bf6b539f68a51aa038f0018864150fdbdc7eac6e","revisions":{"ac490888b789caf621175b4c5e74fa16b6dff07c":{"kind":"REWORK","_number":1,"created":"2022-02-01 15:31:38.000000000","uploader":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"ref":"refs/changes/07/827307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/07/827307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/827307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/827307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/827307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/07/827307/1"}}},"commit":{"parents":[{"commit":"c1969caf119dabff0327b6399712ce226bee570a","subject":"Merge \"Mirror charm-nova-compute-nvidia-vgpu to GitHub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1969caf119dabff0327b6399712ce226bee570a"}]}],"author":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2022-02-01 15:31:25.000000000","tz":60},"committer":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2022-02-01 15:31:25.000000000","tz":60},"subject":"Add dummy charm project","message":"Add dummy charm project\n\nThe default git branch in the upstream project is\n`main` instead of `master. The goal here is to verify\nif such a project can successfully be imported.\n\nChange-Id: I5e4db08f9d7600f7ea045907813d4fb0f6264436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac490888b789caf621175b4c5e74fa16b6dff07c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac490888b789caf621175b4c5e74fa16b6dff07c"}]},"branch":"refs/heads/master"},"14778fc27b80966d7da91b8651788130dd2bbe61":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-01 15:47:23.000000000","uploader":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"ref":"refs/changes/07/827307/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/07/827307/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/827307/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/827307/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/827307/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/07/827307/2"}}},"commit":{"parents":[{"commit":"c1969caf119dabff0327b6399712ce226bee570a","subject":"Merge \"Mirror charm-nova-compute-nvidia-vgpu to GitHub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1969caf119dabff0327b6399712ce226bee570a"}]}],"author":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2022-02-01 15:31:25.000000000","tz":60},"committer":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2022-02-01 15:47:14.000000000","tz":60},"subject":"Add dummy charm project","message":"Add dummy charm project\n\nThe default git branch in the upstream project is\n`main` instead of `master. The goal here is to verify\nif such a project can successfully be imported.\n\nChange-Id: I5e4db08f9d7600f7ea045907813d4fb0f6264436\nRequired-By: I24e21c78b4424b974d1f55c57ff62f91820bb66b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/14778fc27b80966d7da91b8651788130dd2bbe61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/14778fc27b80966d7da91b8651788130dd2bbe61"}]},"branch":"refs/heads/master"},"bf6b539f68a51aa038f0018864150fdbdc7eac6e":{"kind":"REWORK","_number":3,"created":"2022-02-01 15:50:40.000000000","uploader":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"ref":"refs/changes/07/827307/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/07/827307/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/827307/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/827307/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/827307/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/07/827307/3"}}},"commit":{"parents":[{"commit":"c1969caf119dabff0327b6399712ce226bee570a","subject":"Merge \"Mirror charm-nova-compute-nvidia-vgpu to GitHub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1969caf119dabff0327b6399712ce226bee570a"}]}],"author":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2022-02-01 15:31:25.000000000","tz":60},"committer":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2022-02-01 15:50:29.000000000","tz":60},"subject":"Add dummy charm project","message":"Add dummy charm project\n\nThe default git branch in the upstream project is\n`main` instead of `master. The goal here is to verify\nif such a project can successfully be imported.\n\nChange-Id: I5e4db08f9d7600f7ea045907813d4fb0f6264436\nRequired-By: I24e21c78b4424b974d1f55c57ff62f91820bb66b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bf6b539f68a51aa038f0018864150fdbdc7eac6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bf6b539f68a51aa038f0018864150fdbdc7eac6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
