)]}'
{"id":"openstack%2Fproject-config~828435","triplet_id":"openstack%2Fproject-config~master~Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3","project":"openstack/project-config","branch":"master","topic":"rocky-nodepool","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-02-14 21:15:56.000000000","reason":"Clark Boylan replied on the change"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-02-17 13:24:22.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"},"34411":{"account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"last_update":"2022-02-17 14:41:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3","subject":"Add rockylinux-8 to nodepool configuration","status":"MERGED","created":"2022-02-08 22:57:17.000000000","updated":"2022-02-17 14:52:10.000000000","submitted":"2022-02-17 14:41:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828435-rocky-nodepool","meta_rev_id":"616538dbdb548d7b873a699cb6a30210199ae421","_number":828435,"virtual_id_number":828435,"owner":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-17 14:41:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2022-02-17 14:28:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2022-02-17 14:28:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 23:02:08.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2022-02-08 23:08:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 13:19:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-02-14 21:07:11.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-02-15 20:43:19.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2022-02-16 11:18:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"6e71c7cd45d47eea701c4c2e4eca42a28e60deb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-08 22:57:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a32fe534a9562ba23a8d1d80fa0d7d63df3ccf39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-08 22:58:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c755798365639f0cc9ab6ad8f4ff57850b68b029","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-02-08 23:02:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2d69197275216e6870ec441d9fdfd59cf926a4f1","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-02-08 23:03:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"492942053d64dc98d8dc3655f9bb468b14aa161a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 23:08:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3b3dc0e1c4c2448093e09389f0119016 : SUCCESS in 7m 13s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/59247ab2194e463bb35a1f4748063389 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"c6e0ae459f50ca889258312d3a2d2cb38dd3812e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-08 23:20:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b64d6fe43a3d7babf1e18232232979bc6f83e6b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-08 23:25:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"38428bf3d114ea4f40b4d586a9edeedc2d4f936d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-08 23:28:38.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"3b179f548117d79bd32c80c8029a720b2043bf9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 23:28:38.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"76c2da2e2c5f2ce6d3e3ba4d993aca5db0a12cbe","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-08 23:30:41.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5821c5030029f46afeaaded8637610826af5dd32","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-08 23:31:20.000000000","message":"Topic set to support-rockylinux-8","accounts_in_message":[],"_revision_number":5},{"id":"bf2be7cb84045d272090f74abe10cb8ef316b536","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-08 23:32:24.000000000","message":"Topic support-rockylinux-8 removed","accounts_in_message":[],"_revision_number":5},{"id":"fc38eece62ce56a10385e8bde33625246bb81595","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-08 23:32:40.000000000","message":"Topic set to rockylinux-8","accounts_in_message":[],"_revision_number":5},{"id":"3e0d7ae63897c5609b9a160c113863e2ceb1c9ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 23:40:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8add4d8ecd4f43819c42020700eef022 : SUCCESS in 7m 37s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/8908d485577e46aa806ae4e71e2775fc : SUCCESS in 4m 36s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/c12a6f0d8d8d4759abc8583ceafcac17 : FAILURE in 5m 14s","accounts_in_message":[],"_revision_number":5},{"id":"5b78dc7477909da6389e9058c7aa64bf8cd6811d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-09 00:32:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c51900e9721fa6c329269136219c0267910e67f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 00:42:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/df046e29ce8341718f0451740b33fbe4 : SUCCESS in 7m 35s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/104a94eab6484055b77b6b74263ed34a : SUCCESS in 4m 33s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/af542ce7e09e40fa9d0bde6b3f3fcab1 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":6},{"id":"c3739946add344ef8d60125c33f3a034f0aaba32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-11 15:34:53.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7c4e1a305d3ff952731799bbc59b9687e269a715","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 15:44:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a956ab97c2794b14bdef3a79416590bd : SUCCESS in 6m 30s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/0a15d7d9e3854ab5950ec0d31f625689 : SUCCESS in 4m 15s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/bd207c665b9c4f2cb30298a1d2dceb4f : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":7},{"id":"6ee4bd43e5bc2bef47128238fd8351b432a244e5","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-02-14 21:07:11.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"7ebbef4eac4670c6481cbc32458dab33344242e0","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-14 21:08:24.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"fda46edbc14d0a6bf4db72af818a8bd89a6c7440","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-14 21:13:16.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9bf271a8e02b86e1f1e1af7b4fd8c2b2519d54b5","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-02-14 21:15:56.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"0c6699719990b978610c7898c8771c59864c990e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 21:24:49.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/43b7eb95710d45f1842e2d3c874c3c36 : SUCCESS in 7m 03s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/f26d5602556c4096aa43ee20a25276b5 : SUCCESS in 3m 59s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b658ab88aa49476d86d110071ec8dd28 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":8},{"id":"92c7586a43e67fbfab91dc1aca2d3a5d745e7d47","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-02-15 20:43:19.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"0dbb2fec4469a30e1f5b516646f39b71ec9ab608","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-02-15 20:46:54.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"10e5dd5c1be010fe432bc288734900286c31a3ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-15 21:06:05.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"0f8e7b1a7afb9a30ca4ebb32f3ff1534329f1775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 21:10:06.000000000","message":"Patch Set 9: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"60cdbc425cc867fc8d4ae9c8c5600a732bbb6002","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-16 00:36:07.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"27594ebf2e4c32235e649c5bfd0bb97b2c68883e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 00:46:12.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/be5cf2058abc4e09a401e8f3327bf18a : SUCCESS in 7m 32s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/a24ddd9778404e39bf5e2f66613c7f6f : SUCCESS in 6m 24s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/120d0b54b863458c937ddacffb7ca130 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":9},{"id":"bd3dd54e98f5c822ea26fbe39d5ea3b1e9c65e65","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-02-16 11:18:18.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"91e829534bcf0714060716694b4e96d11eab5c9c","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-16 14:31:34.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5c77fd251640cf0997f71ff35575d42c5e69863f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-16 14:33:29.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"04617c8a343e6d3683ac5679fa638a939a1301a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 14:43:02.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/df6b48f07daa46f7860cabaa251d9d59 : SUCCESS in 6m 51s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/8e9a1aaf06cb4d6da365c5553d043f4c : SUCCESS in 4m 53s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/2a91f228be2e43808a9e5f71495e3066 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":10},{"id":"81c5da5fb18dc558c14cf2739b6a8ecf41896159","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-02-16 16:16:01.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"cc741aed7b5a765376f858139373356dcdf07abd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-02-17 13:24:22.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"07afd372d0c419cdb7264df980168af1ed3ece22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 13:24:49.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"797e3d77215799c238d797c696ba1706e57ca44f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 13:33:16.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/230948fb8d48439faef1669dd9771f27 : SUCCESS in 7m 44s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/cf6a7c769b614efea6a20a6dea58da68 : FAILURE in 3m 56s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/495447c4840d4e2d9e687a5386c949e8 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":10},{"id":"06cf3190a85ad60171b493da9af2c8d8cdbe3de6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-17 13:51:10.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"df851741966c550a448c4cc8ac0077223e7dca6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-17 14:00:19.000000000","message":"Uploaded patch set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"5f556a1cc460c3ec55f24f84fc0542841d5c398e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:10:42.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d0abbcb24a6c4946bb4aefb9600ec795 : SUCCESS in 6m 53s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/aa884adea45348589ac989b659f41c6d : FAILURE in 4m 11s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/58a36aa7910e449994fbf550f22e61f1 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":12},{"id":"6ed517792474be56ea619d148e0241f3e01bb632","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-02-17 14:16:12.000000000","message":"Uploaded patch set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"979c789f9e95ac173702b5d9a4fe0e3d8dd0dd15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:25:08.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0e96bb9eed534fc1afc99f5f230f574a : SUCCESS in 7m 02s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/d19546b387c348c3aa3ba49f5d506474 : SUCCESS in 4m 36s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/0193ea7796b24290b7fa3e90a23019e5 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":13},{"id":"c15d1647dc22385e8072029efea1b87083635024","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-02-17 14:28:39.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"859e61e109546ad22977b0358111dcdb23132967","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:29:14.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"80ce45adb3cd32cf2cde695aafbc86a633a137a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:41:09.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dceca9b238174558844eb53e12493f01 : SUCCESS in 7m 44s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/d0d15227c378478da173c1a61ad80166 : SUCCESS in 4m 09s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b34b42af85d84a5ca00b4a16fe4b769b : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":13},{"id":"93670b28ce7e8acc39157a2adbeb86736d455e31","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:41:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"616538dbdb548d7b873a699cb6a30210199ae421","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:52:10.000000000","message":"Patch Set 13:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/92150b7e8905420b8e21c1b8ec580384 : SUCCESS in 6m 10s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/ee70633ca5ce4778b9d3c5180cef3cdd : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"2ccc5241c839f333c08d7feef1fd4da13da729f6","revisions":{"59f26ed9909c967edaf301f27db68c6f3a94b589":{"kind":"REWORK","_number":1,"created":"2022-02-08 22:57:17.000000000","uploader":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"ref":"refs/changes/35/828435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/828435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/828435/1"}}},"commit":{"parents":[{"commit":"efa89519ff7087d28dcd3e13961f12cceee38995","subject":"Merge \"Retire security-specs repo - Step 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efa89519ff7087d28dcd3e13961f12cceee38995"}]}],"author":{"name":"Neil Hanlon","email":"neil@rockylinux.org","date":"2022-02-08 22:56:23.000000000","tz":-300},"committer":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-02-08 22:57:11.000000000","tz":-300},"subject":"Add rockylinux-8 test node","message":"Add rockylinux-8 test node\n\nSigned-off-by: Neil Hanlon \u003cneil@rockylinux.org\u003e\nChange-Id: Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/59f26ed9909c967edaf301f27db68c6f3a94b589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/59f26ed9909c967edaf301f27db68c6f3a94b589"}]},"branch":"refs/heads/master"},"2d721219c8e315e684ee62bfef326486b3de4396":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-08 22:58:28.000000000","uploader":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"ref":"refs/changes/35/828435/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/828435/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/828435/2"}}},"commit":{"parents":[{"commit":"efa89519ff7087d28dcd3e13961f12cceee38995","subject":"Merge \"Retire security-specs repo - Step 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efa89519ff7087d28dcd3e13961f12cceee38995"}]}],"author":{"name":"Neil Hanlon","email":"neil@rockylinux.org","date":"2022-02-08 22:56:23.000000000","tz":-300},"committer":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-02-08 22:58:10.000000000","tz":-300},"subject":"Add rockylinux-8 test node","message":"Add rockylinux-8 test node\n\nSigned-off-by: Neil Hanlon \u003cneil@rockylinux.org\u003e\nChange-Id: Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3\nDepends-On: https://review.opendev.org/c/opendev/base-jobs/+/828436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2d721219c8e315e684ee62bfef326486b3de4396"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2d721219c8e315e684ee62bfef326486b3de4396"}]},"branch":"refs/heads/master"},"a8cf022fa5042c92acdee84a464dc6ec7b2d2a52":{"kind":"REWORK","_number":3,"created":"2022-02-08 23:20:14.000000000","uploader":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"ref":"refs/changes/35/828435/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/828435/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/828435/3"}}},"commit":{"parents":[{"commit":"efa89519ff7087d28dcd3e13961f12cceee38995","subject":"Merge \"Retire security-specs repo - Step 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efa89519ff7087d28dcd3e13961f12cceee38995"}]}],"author":{"name":"Neil Hanlon","email":"neil@rockylinux.org","date":"2022-02-08 22:56:23.000000000","tz":-300},"committer":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-02-08 23:19:53.000000000","tz":-300},"subject":"Add rockylinux-8 to nodepool configuration","message":"Add rockylinux-8 to nodepool configuration\n\nSigned-off-by: Neil Hanlon \u003cneil@rockylinux.org\u003e\nChange-Id: Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3\nDepends-On: https://review.opendev.org/c/opendev/base-jobs/+/828436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a8cf022fa5042c92acdee84a464dc6ec7b2d2a52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a8cf022fa5042c92acdee84a464dc6ec7b2d2a52"}]},"branch":"refs/heads/master"},"91f7fdccbfde9c5cdef4e6577f9ae8a6658a016e":{"kind":"REWORK","_number":4,"created":"2022-02-08 23:25:51.000000000","uploader":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"ref":"refs/changes/35/828435/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/828435/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/828435/4"}}},"commit":{"parents":[{"commit":"efa89519ff7087d28dcd3e13961f12cceee38995","subject":"Merge \"Retire security-specs repo - Step 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efa89519ff7087d28dcd3e13961f12cceee38995"}]}],"author":{"name":"Neil Hanlon","email":"neil@rockylinux.org","date":"2022-02-08 22:56:23.000000000","tz":-300},"committer":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-02-08 23:25:46.000000000","tz":-300},"subject":"Add rockylinux-8 to nodepool configuration","message":"Add rockylinux-8 to nodepool configuration\n\nSigned-off-by: Neil Hanlon \u003cneil@rockylinux.org\u003e\nChange-Id: Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3\nDepends-On: https://review.opendev.org/c/opendev/base-jobs/+/828436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/91f7fdccbfde9c5cdef4e6577f9ae8a6658a016e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/91f7fdccbfde9c5cdef4e6577f9ae8a6658a016e"}]},"branch":"refs/heads/master"},"3f0ba033bf16924c905383358e3cbcc3a14e50a6":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-02-08 23:28:38.000000000","uploader":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"ref":"refs/changes/35/828435/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/828435/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/828435/5"}}},"commit":{"parents":[{"commit":"efa89519ff7087d28dcd3e13961f12cceee38995","subject":"Merge \"Retire security-specs repo - Step 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efa89519ff7087d28dcd3e13961f12cceee38995"}]}],"author":{"name":"Neil Hanlon","email":"neil@rockylinux.org","date":"2022-02-08 22:56:23.000000000","tz":-300},"committer":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-02-08 23:28:29.000000000","tz":-300},"subject":"Add rockylinux-8 to nodepool configuration","message":"Add rockylinux-8 to nodepool configuration\n\nSigned-off-by: Neil Hanlon \u003cneil@rockylinux.org\u003e\nChange-Id: Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3f0ba033bf16924c905383358e3cbcc3a14e50a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3f0ba033bf16924c905383358e3cbcc3a14e50a6"}]},"branch":"refs/heads/master"},"01820819b349380deccb3561d8d30f8d0303376f":{"kind":"REWORK","_number":6,"created":"2022-02-09 00:32:25.000000000","uploader":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"ref":"refs/changes/35/828435/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/828435/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/828435/6"}}},"commit":{"parents":[{"commit":"efa89519ff7087d28dcd3e13961f12cceee38995","subject":"Merge \"Retire security-specs repo - Step 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efa89519ff7087d28dcd3e13961f12cceee38995"}]}],"author":{"name":"Neil Hanlon","email":"neil@rockylinux.org","date":"2022-02-08 22:56:23.000000000","tz":-300},"committer":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-02-09 00:32:22.000000000","tz":-300},"subject":"Add rockylinux-8 to nodepool configuration","message":"Add rockylinux-8 to nodepool configuration\n\nSigned-off-by: Neil Hanlon \u003cneil@rockylinux.org\u003e\nChange-Id: Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/01820819b349380deccb3561d8d30f8d0303376f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/01820819b349380deccb3561d8d30f8d0303376f"}]},"branch":"refs/heads/master"},"07f75803aa138786186f4e881f6e9ebdf1465778":{"kind":"REWORK","_number":7,"created":"2022-02-11 15:34:53.000000000","uploader":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"ref":"refs/changes/35/828435/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/828435/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/828435/7"}}},"commit":{"parents":[{"commit":"e14c59a6e174f8f7b96df8bdc72c2a200c1c1f49","subject":"Merge \"Mirror charm-ops-interface-* to GitHub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c59a6e174f8f7b96df8bdc72c2a200c1c1f49"}]}],"author":{"name":"Neil Hanlon","email":"neil@rockylinux.org","date":"2022-02-08 22:56:23.000000000","tz":-300},"committer":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-02-11 15:34:46.000000000","tz":-300},"subject":"Add rockylinux-8 to nodepool configuration","message":"Add rockylinux-8 to nodepool configuration\n\nSigned-off-by: Neil Hanlon \u003cneil@rockylinux.org\u003e\nChange-Id: Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/07f75803aa138786186f4e881f6e9ebdf1465778"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/07f75803aa138786186f4e881f6e9ebdf1465778"}]},"branch":"refs/heads/master"},"6a7e1abdfe762110c8d915934b373fe6560515d5":{"kind":"REWORK","_number":8,"created":"2022-02-14 21:13:16.000000000","uploader":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"ref":"refs/changes/35/828435/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/828435/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/828435/8"}}},"commit":{"parents":[{"commit":"e14c59a6e174f8f7b96df8bdc72c2a200c1c1f49","subject":"Merge \"Mirror charm-ops-interface-* to GitHub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c59a6e174f8f7b96df8bdc72c2a200c1c1f49"}]}],"author":{"name":"Neil Hanlon","email":"neil@rockylinux.org","date":"2022-02-08 22:56:23.000000000","tz":-300},"committer":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-02-14 21:13:07.000000000","tz":-300},"subject":"Add rockylinux-8 to nodepool configuration","message":"Add rockylinux-8 to nodepool configuration\n\nSigned-off-by: Neil Hanlon \u003cneil@rockylinux.org\u003e\nChange-Id: Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6a7e1abdfe762110c8d915934b373fe6560515d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6a7e1abdfe762110c8d915934b373fe6560515d5"}]},"branch":"refs/heads/master"},"62a691bb1c7f8eae4189b7b2f525d22927e5fca7":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2022-02-15 21:06:05.000000000","uploader":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"ref":"refs/changes/35/828435/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/828435/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/828435/9"}}},"commit":{"parents":[{"commit":"e14c59a6e174f8f7b96df8bdc72c2a200c1c1f49","subject":"Merge \"Mirror charm-ops-interface-* to GitHub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c59a6e174f8f7b96df8bdc72c2a200c1c1f49"}]}],"author":{"name":"Neil Hanlon","email":"neil@rockylinux.org","date":"2022-02-08 22:56:23.000000000","tz":-300},"committer":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-02-15 21:05:55.000000000","tz":-300},"subject":"Add rockylinux-8 to nodepool configuration","message":"Add rockylinux-8 to nodepool configuration\n\nSigned-off-by: Neil Hanlon \u003cneil@rockylinux.org\u003e\nChange-Id: Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/829405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62a691bb1c7f8eae4189b7b2f525d22927e5fca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62a691bb1c7f8eae4189b7b2f525d22927e5fca7"}]},"branch":"refs/heads/master"},"7b023679cb01a715c7c794a4b38197e5783579ba":{"kind":"REWORK","_number":10,"created":"2022-02-16 14:33:29.000000000","uploader":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"ref":"refs/changes/35/828435/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/828435/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/828435/10"}}},"commit":{"parents":[{"commit":"e14c59a6e174f8f7b96df8bdc72c2a200c1c1f49","subject":"Merge \"Mirror charm-ops-interface-* to GitHub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c59a6e174f8f7b96df8bdc72c2a200c1c1f49"}]}],"author":{"name":"Neil Hanlon","email":"neil@rockylinux.org","date":"2022-02-08 22:56:23.000000000","tz":-300},"committer":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-02-16 14:33:12.000000000","tz":-300},"subject":"Add rockylinux-8 to nodepool configuration","message":"Add rockylinux-8 to nodepool configuration\n\nSigned-off-by: Neil Hanlon \u003cneil@rockylinux.org\u003e\nChange-Id: Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/829405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b023679cb01a715c7c794a4b38197e5783579ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b023679cb01a715c7c794a4b38197e5783579ba"}]},"branch":"refs/heads/master"},"5b78acf583e9c1868a2846079b8a3d60bc148fed":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2022-02-17 13:51:10.000000000","uploader":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"ref":"refs/changes/35/828435/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/828435/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/828435/11"}}},"commit":{"parents":[{"commit":"e14c59a6e174f8f7b96df8bdc72c2a200c1c1f49","subject":"Merge \"Mirror charm-ops-interface-* to GitHub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c59a6e174f8f7b96df8bdc72c2a200c1c1f49"}]}],"author":{"name":"Neil Hanlon","email":"neil@rockylinux.org","date":"2022-02-08 22:56:23.000000000","tz":-300},"committer":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-02-17 13:49:35.000000000","tz":-300},"subject":"Add rockylinux-8 to nodepool configuration","message":"Add rockylinux-8 to nodepool configuration\n\nSigned-off-by: Neil Hanlon \u003cneil@rockylinux.org\u003e\nChange-Id: Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/829405\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/829712\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5b78acf583e9c1868a2846079b8a3d60bc148fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5b78acf583e9c1868a2846079b8a3d60bc148fed"}]},"branch":"refs/heads/master"},"e59bebc0d65b710edf6a89fa62369519ce2ccd80":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2022-02-17 14:00:19.000000000","uploader":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"ref":"refs/changes/35/828435/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/828435/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/828435/12"}}},"commit":{"parents":[{"commit":"e14c59a6e174f8f7b96df8bdc72c2a200c1c1f49","subject":"Merge \"Mirror charm-ops-interface-* to GitHub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c59a6e174f8f7b96df8bdc72c2a200c1c1f49"}]}],"author":{"name":"Neil Hanlon","email":"neil@rockylinux.org","date":"2022-02-08 22:56:23.000000000","tz":-300},"committer":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-02-17 13:59:48.000000000","tz":-300},"subject":"Add rockylinux-8 to nodepool configuration","message":"Add rockylinux-8 to nodepool configuration\n\nSigned-off-by: Neil Hanlon \u003cneil@rockylinux.org\u003e\nChange-Id: Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/829405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e59bebc0d65b710edf6a89fa62369519ce2ccd80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e59bebc0d65b710edf6a89fa62369519ce2ccd80"}]},"branch":"refs/heads/master"},"2ccc5241c839f333c08d7feef1fd4da13da729f6":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2022-02-17 14:16:12.000000000","uploader":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"ref":"refs/changes/35/828435/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/828435/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/828435/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/828435/13"}}},"commit":{"parents":[{"commit":"e14c59a6e174f8f7b96df8bdc72c2a200c1c1f49","subject":"Merge \"Mirror charm-ops-interface-* to GitHub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c59a6e174f8f7b96df8bdc72c2a200c1c1f49"}]}],"author":{"name":"Neil Hanlon","email":"neil@rockylinux.org","date":"2022-02-08 22:56:23.000000000","tz":-300},"committer":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-02-17 14:12:00.000000000","tz":-300},"subject":"Add rockylinux-8 to nodepool configuration","message":"Add rockylinux-8 to nodepool configuration\n\nSigned-off-by: Neil Hanlon \u003cneil@rockylinux.org\u003e\nChange-Id: Ic3344bc47ca56c27f7ec3427a0865bd6ce3349d3\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/829405\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/829712\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2ccc5241c839f333c08d7feef1fd4da13da729f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2ccc5241c839f333c08d7feef1fd4da13da729f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
