)]}'
{"id":"openstack%2Fproject-config~831398","triplet_id":"openstack%2Fproject-config~master~Icfb8fc6d2b15714ecb58960d8e44b199bedd6b0d","project":"openstack/project-config","branch":"master","topic":"remove-airship-citycloud","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-03-01 18:14:06.000000000","reason":"Clark Boylan replied on the change"}},"hashtags":[],"change_id":"Icfb8fc6d2b15714ecb58960d8e44b199bedd6b0d","subject":"Remove airship-citycloud nodepool provider","status":"MERGED","created":"2022-03-01 15:57:30.000000000","updated":"2022-03-01 18:53:21.000000000","submitted":"2022-03-01 18:23:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831398-remove-airship-citycloud","meta_rev_id":"115bf6b070b13bb5a30f5761733f0647b6993694","_number":831398,"virtual_id_number":831398,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-01 18:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2022-03-01 16:21:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-01 16:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2022-03-01 18:14:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-01 16:16:33.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-03-01 16:21:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-03-01 16:36:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6878407311e9a4ba61c1e9d7eda4a93e6c7c84f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-03-01 15:57:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7852dd94c13fe473b0d51dc080ab80694d058900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-03-01 16:11:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"224e21b496dacc4f346a6afa470dbbaaf2cb0883","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-03-01 16:16:33.000000000","message":"Patch Set 2: Code-Review+2\n\nThis should also be safe to merge immediately, since it just stops management of the resources and will help us avoid our deployment jobs breaking if the server goes offline before we\u0027re done.","accounts_in_message":[],"_revision_number":2},{"id":"9fdcb3817b649521b3df606243833e34e1b6b36a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-03-01 16:18:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c03f2cceef046a5c64bb8bd82f56753d5266f3f7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-03-01 16:21:52.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7b2137e5db9cd651c665dc0b8eef534e219aadee","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-03-01 16:22:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0831b53380be017c972e36a79161d0b45f5a357a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-03-01 16:23:49.000000000","message":"Patch Set 3: Code-Review+2\n\nOh, right, per discussion in IRC this removes the nodepool configuration, which we don\u0027t want to do until all the nodes and images have drained.","accounts_in_message":[],"_revision_number":3},{"id":"bdd7fce9c17b3a30e0f304c5782e9d036c7949f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 16:36:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e83f13ab5e7543b0adf0514ef028e235 : SUCCESS in 8m 13s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/64f9b79e0da942a28b411b5a6930926d : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":3},{"id":"4b33ac982fe6ffc787d613fe2f2cdb1e5158f30e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-03-01 18:14:06.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"87aded30208b1d442968358bfd7435224d507f77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 18:14:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6f568fb05e1e8ab438d8b62504ef39cc19e3e31c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 18:23:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e4410f236c314537b640bff0e80ecd5e : SUCCESS in 8m 20s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/9cea06ebaa2e459487d1120e318cff9c : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":3},{"id":"0cb4f5de952af6f938c25c2e65e588c39011c7e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 18:23:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"115bf6b070b13bb5a30f5761733f0647b6993694","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 18:53:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/8341a378b1f74f849858b2b9bd006772 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3dc09d41d1ea178e13096ff6b0e424d22e78adbc","revisions":{"a9f5cbcdde46c3611846ca2bc433faa697db7b35":{"kind":"REWORK","_number":1,"created":"2022-03-01 15:57:30.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/98/831398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/98/831398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/831398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/831398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/831398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/98/831398/1"}}},"commit":{"parents":[{"commit":"d4d0c5b5527290faed2cd46455ecbccf955bfb19","subject":"Set airship-citycloud max-servers to 0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d4d0c5b5527290faed2cd46455ecbccf955bfb19"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-03-01 15:56:08.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-03-01 15:56:08.000000000","tz":-480},"subject":"Remove airship-citycloud nodepool provider","message":"Remove airship-citycloud nodepool provider\n\nOnce we\u0027ve settled on a max-servers value of 0 without any servers this\nis the nextstep in removing the provider. This should completely remove\nthe provider from the launcher and the image builders.\n\nChange-Id: Icfb8fc6d2b15714ecb58960d8e44b199bedd6b0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a9f5cbcdde46c3611846ca2bc433faa697db7b35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a9f5cbcdde46c3611846ca2bc433faa697db7b35"}]},"branch":"refs/heads/master"},"6a450f4f235594b48d4a1cddcdb157b51ed93929":{"kind":"REWORK","_number":2,"created":"2022-03-01 16:11:25.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/98/831398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/98/831398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/831398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/831398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/831398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/98/831398/2"}}},"commit":{"parents":[{"commit":"1c73cd826c1decea3c2abbfb7f152102440cdefe","subject":"Set airship-citycloud max-servers to 0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1c73cd826c1decea3c2abbfb7f152102440cdefe"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-03-01 15:56:08.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-03-01 16:11:05.000000000","tz":-480},"subject":"Remove airship-citycloud nodepool provider","message":"Remove airship-citycloud nodepool provider\n\nOnce we\u0027ve settled on a max-servers value of 0 without any servers this\nis the nextstep in removing the provider. This should completely remove\nthe provider from the launcher and the image builders.\n\nChange-Id: Icfb8fc6d2b15714ecb58960d8e44b199bedd6b0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6a450f4f235594b48d4a1cddcdb157b51ed93929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6a450f4f235594b48d4a1cddcdb157b51ed93929"}]},"branch":"refs/heads/master"},"3dc09d41d1ea178e13096ff6b0e424d22e78adbc":{"kind":"REWORK","_number":3,"created":"2022-03-01 16:18:55.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/98/831398/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/98/831398/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/831398/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/831398/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/831398/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/98/831398/3"}}},"commit":{"parents":[{"commit":"1c73cd826c1decea3c2abbfb7f152102440cdefe","subject":"Set airship-citycloud max-servers to 0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1c73cd826c1decea3c2abbfb7f152102440cdefe"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-03-01 15:56:08.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-03-01 16:18:08.000000000","tz":-480},"subject":"Remove airship-citycloud nodepool provider","message":"Remove airship-citycloud nodepool provider\n\nOnce we\u0027ve settled on a max-servers value of 0 without any servers this\nis the nextstep in removing the provider. This should completely remove\nthe provider from the launcher and the image builders.\n\nWe keep the airship-citycloud nodepool provider for historical\ninformation purposes. We can clean this up later.\n\nChange-Id: Icfb8fc6d2b15714ecb58960d8e44b199bedd6b0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3dc09d41d1ea178e13096ff6b0e424d22e78adbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3dc09d41d1ea178e13096ff6b0e424d22e78adbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
