)]}'
{"id":"openstack%2Fproject-config~834658","triplet_id":"openstack%2Fproject-config~master~I5a6958175bd2c4c0fec637a391672ccefcd22c74","project":"openstack/project-config","branch":"master","topic":"retirement","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-03-28 10:46:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5a6958175bd2c4c0fec637a391672ccefcd22c74","subject":"Initiate retirement of openstack/os-testr","status":"ABANDONED","created":"2022-03-22 11:41:43.000000000","updated":"2022-03-28 10:46:11.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ed9c6fb0efd90ce2e2537b31d9fc25f4e83ed07c","_number":834658,"virtual_id_number":834658,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-22 11:57:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":-1,"date":"2022-03-22 14:27:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-22 11:57:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-22 14:27:52.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"13f735e25c2c424ecc9a2a78aeb701b8cddff314","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-22 11:41:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"469c6bd727f91441f2f31213ef7d028c9df204af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 11:57:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9d4f141479854e5ab55d49f22a52043d : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":1},{"id":"507c08ffe6eddbeb2b45fd2ae4c8e941563588f5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-03-22 14:27:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed9c6fb0efd90ce2e2537b31d9fc25f4e83ed07c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-28 10:46:11.000000000","message":"Abandoned\n\nThe ostestr command has been retired instead https://review.opendev.org/c/openstack/os-testr/+/835124/2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f616a9f620c146b9ed2a5ae3ce9aefefeceddd49","revisions":{"f616a9f620c146b9ed2a5ae3ce9aefefeceddd49":{"kind":"REWORK","_number":1,"created":"2022-03-22 11:41:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/58/834658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/58/834658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/834658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/834658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/834658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/58/834658/1"}}},"commit":{"parents":[{"commit":"9e78306b15a9df4df5b8aa14b1ce9b12d95a0cc7","subject":"grafana afs : remove centos 8 volume tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9e78306b15a9df4df5b8aa14b1ce9b12d95a0cc7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-03-22 11:35:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-03-22 11:35:16.000000000","tz":0},"subject":"Initiate retirement of openstack/os-testr","message":"Initiate retirement of openstack/os-testr\n\nAs announced on the mailing list [1].\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2022-March/027784.html\n\nChange-Id: I5a6958175bd2c4c0fec637a391672ccefcd22c74\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f616a9f620c146b9ed2a5ae3ce9aefefeceddd49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f616a9f620c146b9ed2a5ae3ce9aefefeceddd49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
