)]}'
{"id":"openstack%2Fproject-config~834699","triplet_id":"openstack%2Fproject-config~master~I17d3a4aebb0670aa14885449050160b44bea4e8b","project":"openstack/project-config","branch":"master","topic":"retire-puppet-tuskar","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-24 19:39:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I17d3a4aebb0670aa14885449050160b44bea4e8b","subject":"Formally retire puppet-tuskar","status":"MERGED","created":"2022-03-22 15:04:17.000000000","updated":"2022-08-24 21:18:00.000000000","submitted":"2022-08-24 19:39:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834699-retire-puppet-tuskar","meta_rev_id":"7c893139756149df3cb1d096d7c6ce7e0cce0bfb","_number":834699,"virtual_id_number":834699,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-24 19:39:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-24 19:31:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-24 19:31:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-22 15:20:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-24 19:31:26.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"1995841e6c32c9bdc2967cd18dde8a226e371bcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-22 15:04:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1406d759d93c99d68324e56fe12d35f932b235b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 15:20:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/095bcb3274994f8081be7fdf5882119b : SUCCESS in 8m 17s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ab23f449ba354d17aef3f9482cf2dcf5 : SUCCESS in 5m 47s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/6ccc6fc7f65d45aba66c4a4b6e2a28fa : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ac8ea8e5f551985eecd1cde72792aa17077d36ef","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-24 19:31:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f75c7750c2970258ed8dfe9442bf53dc9a631574","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 19:31:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f471ef21abe5faa833846d452648d65b5c2c04e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 19:39:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20cacaf0fb0e48d1976e83a2c0176f1f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/89f4314c6ecc416b99bc094539eeefc9 : SUCCESS in 6m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3f296f50780b484d90b3d3c3f2ecd5b6 : SUCCESS in 4m 27s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/1dbee4cb4fe14ea3960812ad92b307e7 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f1c259128effcd176c70c5540984df7087f53e35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 19:39:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7c893139756149df3cb1d096d7c6ce7e0cce0bfb","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 21:18:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52fb7663c6e14fee90fa2865fe209413\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/a2a4782900a4472bb60e0739708a0a45 : SUCCESS in 8m 28s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/c7880341b96b4d8fa70ff33571aa65c2 : SUCCESS in 9m 08s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/f9eb364147364e038fd15b5a92604508 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f08c3c611ad20564f1a3140d34934cd00ac6fb88","revisions":{"f08c3c611ad20564f1a3140d34934cd00ac6fb88":{"kind":"REWORK","_number":1,"created":"2022-03-22 15:04:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/834699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/99/834699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/834699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/834699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/834699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/99/834699/1"}}},"commit":{"parents":[{"commit":"e091dbcb5abc8aa48bdf07ebf9010574bb9d9356","subject":"Merge \"Finalize batch of opendev repo retirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e091dbcb5abc8aa48bdf07ebf9010574bb9d9356"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-22 14:56:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-22 15:04:01.000000000","tz":540},"subject":"Formally retire puppet-tuskar","message":"Formally retire puppet-tuskar\n\npuppet-tuskar was already retired[1]. This change removes/updates\nthe remaining configuration about that retired repo.\n\n[1] https://review.opendev.org/c/x/puppet-tuskar/+/263919\n\nChange-Id: I17d3a4aebb0670aa14885449050160b44bea4e8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f08c3c611ad20564f1a3140d34934cd00ac6fb88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f08c3c611ad20564f1a3140d34934cd00ac6fb88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
