)]}'
{"id":"openstack%2Fproject-config~835823","triplet_id":"openstack%2Fproject-config~master~I22e89de70e925f984d7b4e60e705a135a179487a","project":"openstack/project-config","branch":"master","topic":"disable-iweb","hashtags":[],"change_id":"I22e89de70e925f984d7b4e60e705a135a179487a","subject":"Revert \"Temporarily disable launching in iWeb\"","status":"ABANDONED","created":"2022-03-30 11:55:05.000000000","updated":"2022-03-30 12:12:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"8b8cab36ae6c684e95a0a0915154859f3d61e3a2","_number":835823,"virtual_id_number":835823,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-30 12:08:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-03-30 12:08:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c7d69bbf5ccd1ec239cdae0b63262ebf5281ec1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-03-30 11:55:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cbab118bbd90858bb49d54c722c8805fd18df7d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-03-30 11:56:39.000000000","message":"Set Work In Progress\n\nHold until we get confirmation it\u0027s safe to resume.","accounts_in_message":[],"_revision_number":1},{"id":"79c3767fb3584ec7147c2964c90cabc6144fa84a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 12:08:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/399a2d89e2064356adc071e66f0dc273 : SUCCESS in 8m 45s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/4ecf11c77405431da48955b3a56dcfc8 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":1},{"id":"8b8cab36ae6c684e95a0a0915154859f3d61e3a2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-03-30 12:12:54.000000000","message":"Abandoned\n\nThe parent change was abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6db421dd75d8651c1fc48daf728656276b17d702","revisions":{"6db421dd75d8651c1fc48daf728656276b17d702":{"kind":"REWORK","_number":1,"created":"2022-03-30 11:55:05.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/23/835823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/835823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/835823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/835823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/835823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/835823/1"}}},"commit":{"parents":[{"commit":"69b3bef0cb2b2168d4b436542e4e94f4ed66a687","subject":"Temporarily disable launching in iWeb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/69b3bef0cb2b2168d4b436542e4e94f4ed66a687"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-03-30 11:54:37.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-03-30 11:54:58.000000000","tz":0},"subject":"Revert \"Temporarily disable launching in iWeb\"","message":"Revert \"Temporarily disable launching in iWeb\"\n\nThis reverts commit 69b3bef0cb2b2168d4b436542e4e94f4ed66a687.\n\nMerge once we get confirmation the situation is resolved.\n\nChange-Id: I22e89de70e925f984d7b4e60e705a135a179487a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6db421dd75d8651c1fc48daf728656276b17d702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6db421dd75d8651c1fc48daf728656276b17d702"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
