)]}'
{"id":"openstack%2Fproject-config~836200","triplet_id":"openstack%2Fproject-config~master~I07ee7ccdd10e68f6f38ec362ce7e5d333d5fc2f7","project":"openstack/project-config","branch":"master","topic":"centos-stream","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-04-04 04:26:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I07ee7ccdd10e68f6f38ec362ce7e5d333d5fc2f7","subject":"Match configure-mirrors for CentOS wheel URLs","status":"MERGED","created":"2022-04-01 16:14:40.000000000","updated":"2022-04-04 04:26:58.000000000","submitted":"2022-04-04 04:26:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836200-centos-stream","meta_rev_id":"cd1d62404870b8b3a44447cedbcf201342230003","_number":836200,"virtual_id_number":836200,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-04 04:26:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-04-01 16:15:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-04 04:17:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-04 04:17:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-01 16:15:52.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-04-01 16:23:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-04 04:17:56.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"18ced8e76d4d23d9ce9369ac1e27f1a7658d3488","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-04-01 16:14:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ee726503075a13de052c84c374eff611ca9e505","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-04-01 16:15:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9491f4c5aa9724b676fe1daa6de4734935e9a468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 16:23:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c7dcfe750d3c46c7adb203c327869160 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a8f4bef40f2f7ae2609c419dcbd3357c7b98b3fc","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-04-04 04:17:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e5678d68a607654ac9a43d660332d219a2a2b26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 04:18:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a902f0fe49cc06e74c343dc1eecb8e00aaf638a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 04:26:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e05c42f6be504eb89e882d779a278d92 : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":1},{"id":"cd1d62404870b8b3a44447cedbcf201342230003","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 04:26:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17aff0721bda82d00409fcccd2c367b850395226","revisions":{"17aff0721bda82d00409fcccd2c367b850395226":{"kind":"REWORK","_number":1,"created":"2022-04-01 16:14:40.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/00/836200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/00/836200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/836200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/836200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/836200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/00/836200/1"}}},"commit":{"parents":[{"commit":"6e5c367b2393da173fbd53b41f91b166a4fcd9a1","subject":"Merge \"Revert \"[OVH/GRA1] Disable nodepool temporarily\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6e5c367b2393da173fbd53b41f91b166a4fcd9a1"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-04-01 15:59:52.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-04-01 16:04:57.000000000","tz":0},"subject":"Match configure-mirrors for CentOS wheel URLs","message":"Match configure-mirrors for CentOS wheel URLs\n\nThe configure-mirrors role in zuul-jobs expects CentOS wheels will\nbe in URLs like distro-ver-arch and so doesn\u0027t include \"-stream\"\nlike our wheel publishing jobs are adding. This was a necessary\ndistinction in order to differentiate between CentOS Linux 8 and\nCentOS Stream 8 wheels, but since configure-mirrors didn\u0027t know to\ntell Stream nodes to look in the other location, this was actually\nbroken from the moment it was introduced and we simply never noticed\nuntil we removed the old centos-8 volume (at which point\ncentos-8-stream nodes no longer found any wheels).\n\nAs we no longer have centos-8 nodes, rather than try to change how\nconfigure-mirrors works (since it may be successfully used in other\nsites already and would need advance warning or complex overrides in\nbase jobs), simply rearrange our mirrors to mount the volumes in the\npaths it expects. Once this merges I\u0027ll adjust AFS accordingly.\n\nThis partially reverts 3d776f6734ff0c5197471c7a89aa84511a0eff88.\n\nChange-Id: I07ee7ccdd10e68f6f38ec362ce7e5d333d5fc2f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/17aff0721bda82d00409fcccd2c367b850395226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/17aff0721bda82d00409fcccd2c367b850395226"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
