)]}'
{"id":"openstack%2Fproject-config~836793","triplet_id":"openstack%2Fproject-config~master~I4923b2bfaf7c5bbfa327aa64f9c3ae3fd658e0cf","project":"openstack/project-config","branch":"master","topic":"add-centos-stream-9-stuff","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-11 12:34:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4923b2bfaf7c5bbfa327aa64f9c3ae3fd658e0cf","subject":"Build and publish wheel mirror for CentOS Stream 9","status":"MERGED","created":"2022-04-06 10:24:13.000000000","updated":"2022-04-11 12:34:42.000000000","submitted":"2022-04-11 12:34:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836793-add-centos-stream-9-stuff","meta_rev_id":"b7a32e2c994de218a2153c88e96e82c9abd0c9fb","_number":836793,"virtual_id_number":836793,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-11 12:34:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-11 12:07:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-11 12:21:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-06 10:25:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 11:46:56.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2022-04-11 12:07:19.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"88b6ec3b2aa8c792e3b27c9361dde69bcebd3e64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-06 10:24:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01137d472e0dd9e8abfc48ad758f405e57387822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 10:25:06.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Job build-wheel-cache-centos-9-stream not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: publish-wheel-cache-centos-9-stream\n      description: |\n        Publish CentOS Stream 9 wheels for OpenStack CI mirrors.\n      parent: build-wheel-cache-centos-9-stream\n      pre-run: playbooks/openafs-client/setup.yaml\n      post-run: playbooks/publish/wheel-mirror.yaml\n      allowed-projects:\n        - openstack/requirements\n      final: true\n      secrets:\n        - name: afs\n          secret: wheel_keytab\n\n  in \"openstack/project-config/zuul.d/jobs.yaml@master\", line 896, column 3","accounts_in_message":[],"_revision_number":1},{"id":"461dbce9412a7341a747cafc3fd679cbb3edefc4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-06 11:21:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aca20002c7edd725aaf84785507683715eb38c18","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-04-06 11:46:56.000000000","message":"Patch Set 1:\n\nYou\u0027ll want to define them in openstack/openstack-zuul-jobs zuul.d/jobs.yaml like the others, yes.","accounts_in_message":[],"_revision_number":1},{"id":"b3d7ed774c65d709cf6c3c1bb78f73f321a9c6d7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-04-06 11:48:25.000000000","message":"Patch Set 1:\n\nAnd don\u0027t forget to also push a change to openstack/requirements to run them.","accounts_in_message":[],"_revision_number":1},{"id":"5a71d84c0a2926e2a9cca28ceeee9753264f57fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-06 13:19:40.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"41d28fdbe02e852187fff0be89e55519682a31c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 13:31:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5bbc772c221b413ab70282174223b6d8 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":2},{"id":"1c81903d9f626db38547aa067cd33bc60664a7f3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-06 13:39:40.000000000","message":"Topic set to add-centos-stream-9-stuff","accounts_in_message":[],"_revision_number":2},{"id":"68b6cd8ca54aae0fea8d0a1056b7d6dcb6b817ae","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-04-06 13:43:44.000000000","message":"Patch Set 2:\n\nLGTM, but an AFS admin will need to create the two new volumes first before this can be safely merged.","accounts_in_message":[],"_revision_number":2},{"id":"532c1f878b9ca74cf862eda2fba4af1970bd8ace","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-04-11 12:07:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"13711f4c557fef7fc7b95b223e0d14fb394dde08","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-04-11 12:08:22.000000000","message":"Patch Set 2:\n\nThe volumes exist and the dependency has been approved.","accounts_in_message":[],"_revision_number":2},{"id":"05f44e026e777fd01556fae959194a301632ffed","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-04-11 12:21:32.000000000","message":"Patch Set 2: Workflow+1\n\nThe dependency has merged.","accounts_in_message":[],"_revision_number":2},{"id":"2f6c0edb667c9de5f46e19d7f0c4836ac348b88e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:22:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"940e6c82ac77f19b2c2b4c15fb76e25f23d2e206","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:34:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5bf9b9585e734526b212bd3c14db9ee6 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b7a32e2c994de218a2153c88e96e82c9abd0c9fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:34:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e8a32266bd84c429368bedcc7768ecbb024a947","revisions":{"cc55137e439bbc3174e071e9193f4d87bdf034bf":{"kind":"REWORK","_number":1,"created":"2022-04-06 10:24:13.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/93/836793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/93/836793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/836793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/836793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/836793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/93/836793/1"}}},"commit":{"parents":[{"commit":"551b915b8a17cec720ff5959e98456ddcab441a4","subject":"Retire openstack-health project: end project gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/551b915b8a17cec720ff5959e98456ddcab441a4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-06 10:22:46.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-06 10:24:05.000000000","tz":120},"subject":"Build and publish wheel mirror for CentOS Stream 9","message":"Build and publish wheel mirror for CentOS Stream 9\n\nChange-Id: I4923b2bfaf7c5bbfa327aa64f9c3ae3fd658e0cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cc55137e439bbc3174e071e9193f4d87bdf034bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cc55137e439bbc3174e071e9193f4d87bdf034bf"}]},"branch":"refs/heads/master"},"7e8a32266bd84c429368bedcc7768ecbb024a947":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-04-06 13:19:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/93/836793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/93/836793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/836793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/836793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/836793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/93/836793/2"}}},"commit":{"parents":[{"commit":"551b915b8a17cec720ff5959e98456ddcab441a4","subject":"Retire openstack-health project: end project gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/551b915b8a17cec720ff5959e98456ddcab441a4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-06 10:22:46.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-06 13:19:40.000000000","tz":0},"subject":"Build and publish wheel mirror for CentOS Stream 9","message":"Build and publish wheel mirror for CentOS Stream 9\n\nDepends-on: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/836829\n\nChange-Id: I4923b2bfaf7c5bbfa327aa64f9c3ae3fd658e0cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e8a32266bd84c429368bedcc7768ecbb024a947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e8a32266bd84c429368bedcc7768ecbb024a947"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
